)]}'
{"id":"openstack%2Fkolla~872270","triplet_id":"openstack%2Fkolla~master~I5cf944c509f23d430665de458cbbdb2fd31db7bb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-22 14:27:38.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-22 12:46:10.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I5cf944c509f23d430665de458cbbdb2fd31db7bb","subject":"neutron: Add iproute-tc","status":"MERGED","created":"2023-01-31 10:28:37.000000000","updated":"2023-03-22 14:28:47.000000000","submitted":"2023-03-22 14:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872270","meta_rev_id":"4e22366f88286ef92246551b21f6f535ea326bd4","_number":872270,"virtual_id_number":872270,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-22 14:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-03-22 10:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-03-22 10:59:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-22 12:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-22 12:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 11:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 13:32:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-02-27 11:43:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-03-17 08:19:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f499de49600e8d70f660934c7a6e767e6ba4b104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 10:28:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5352d11a4b10f942d82c59ae519a475c83bf41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 11:48:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f53a24ae6794ba7af09143fc69e2d73\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b1e54a9d18074bf0bb523824e57982e4 : SUCCESS in 28m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bd36d28b2ffa4b0ea491554fd59c934c : SUCCESS in 1h 15m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86e386eb33bc41e58e17885ebda099df : FAILURE in 25m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/37a96b20ed784880b34d28fdcace9ba5 : SUCCESS in 57m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/840558c527da4edb957643a096bb5876 : SUCCESS in 1h 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddb68495c1f413a854c18a787598564 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fa7df2fa2c64b019187f3c71a48aa99 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f288123a15f246d6ac90ec744a3a62e8 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebe1520eed4d42d3b4766a91e6bc15fd : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60afdf8c18304a6a8705074d7b5ad536 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca3eabe189fd4bf9982bfeafe55fd531 : SUCCESS in 37m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/18dd3d84a21546ca9c1209eca371ac8c : SUCCESS in 1h 06m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2b2f7de375d94e9bb09472042f9413d0 : FAILURE in 23m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f926227ca60448c09c7cc2a2a7be36ba : SUCCESS in 50m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/323011b058c044c4a7f0dbcab9060858 : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3548516c05a72bd60feb887def11414fdcb122a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 13:47:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0652c0e9957345c886b9859a797f62aa\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b2c216ec5d5d44faa26f0ef023b965e8 : SUCCESS in 40m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/830768c5c8be468dbf7485db6b3d2821 : FAILURE in 23m 15s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/348b765d6d3145338c18ea0a17bbfd9a : SUCCESS in 1h 17m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/161b8d7a58b447dbb602bfa350d56b24 : SUCCESS in 2m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/00cf9aa8543f4584b04ff6b0a211ced1 : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/889bf14f78994b8f9384c2053638ab19 : TIMED_OUT in 3h 15m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/728dc155e6dc4d8792ea8242cf152ac1 : FAILURE in 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ae9a63e4c094ee36bfed6cf07415c31c69609fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-22 13:32:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67929c42d9f0641dba40c9de8ba476468def5b30","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-27 11:43:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b4b714121c2b89f69ebab68ec809300f25fb2b18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 08:19:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d84873d228a6768bc3b7a05fcc999be85191370b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 09:59:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d6325ba889b4120bc8f0e689dcca40b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2ed21e6315bf493b9bc9d8877fbf8c42 : FAILURE in 24m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fe0b8d221ad347759e05bf33d0277f83 : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8e3c25e3a94a409c844348c13db7f0ea : SUCCESS in 18m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c1b32b802e784299b34a1cb9e26c6883 : SUCCESS in 1h 11m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/50269f5f98a748c885d86c3280d5d56e : SUCCESS in 1h 07m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c7e83f408a4912b6876c039afb57c0 : FAILURE in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/000c63d0caff47af825c6e9668e6269a : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f339b5349014480badc3742f9330da5a : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea4e5365db24424bb38abee6860d699f : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4573b3491264157bf1564205437b953 : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/41b13da5a5434182979ff489a5cb08c2 : FAILURE in 31m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f80c657856d6489a8f014a40c2810efd : SUCCESS in 1h 06m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bdc247bd691444ad87a50d2d9233011d : SUCCESS in 1h 06m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7b7548e2ddfa43c48a672402fb18f0d2 : SUCCESS in 18m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/236302edc6744b3bb84f77eae00f432a : SUCCESS in 49m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b2badc7bb6f843cd8bfaac4f0aff99ba : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"79defd2e18c556dc7ba18059ad0981cc53783dba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 10:19:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf5f4651cf14e86b6795080732c92e1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9d5804f327f94278aa54a5b39058c11f : SUCCESS in 16m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e84c9b84416f4da799bfcf56e71c9811 : SUCCESS in 57m 33s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e262518791144733ac27caa3dfa2590f : SUCCESS in 1h 58m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37595e43f8734c75bdf9746dbbddb4ec : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/512dd0b20c7941a3875cf4188d434c9e : SUCCESS in 2m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bd70c2cd2d5c48b0bff34210261a8847 : FAILURE in 1h 02m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/150255be0c134a6c94d4932375b8558d : FAILURE in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87dc6afffad28adc2d3f5ad21b458213fec5b4f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-22 10:59:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2d9454e9de6c11fa8a399aa65b5c147ff15e7432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 12:42:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dae4dc0cf9704efd8dadf646c1a7190d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6f9d1f786b484bd199fee0599226cb23 : SUCCESS in 16m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9f7034e4a34a464b830a51c875084966 : SUCCESS in 1h 05m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89db7da54f974143bf4fb570c467abec : SUCCESS in 17m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/df12c83a8b614735a978e503601f10df : SUCCESS in 1h 13m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b48e8e6d17a44ba98b37918256d25290 : SUCCESS in 1h 23m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64e68143db849b8b1a581f2ef1eede4 : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08b24e49be104262b1921bc4b36474c4 : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f798535c24b1444882ede66d89a2c3a7 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/204cdf48d54b429788e3a7343902f2ca : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9222fd4075a346e2bdcc1fb751cf4880 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc83e1063f044eeab77f24f7c970df0a : SUCCESS in 22m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8503ff6cc3c4405da90f5eac0bf23528 : SUCCESS in 1h 11m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ed6486baf3a6491c99048a1b5af9123d : SUCCESS in 1h 17m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/92ae7238c0ec4731b7c9dcf2877573a4 : SUCCESS in 18m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c642dd2f03384d12bcd17bafd7610f06 : SUCCESS in 48m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/756eb6b3a8cb4abfa3431d9a43a15d3d : SUCCESS in 1h 16m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8e5d0031afa3510cc06441ae8ff24dbdf84cd38e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-22 12:46:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"710673e379600898e27647175f85e6532f670f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 12:46:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c4749d45102464d62e5d019760bfdaa865323548","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 12:59:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11942c38b0ba4a5584774f244c8c9f73\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fd679f22da45441da5f53f0342613cef : SUCCESS in 51m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a4492c7def2649739e9b272b6bc2ce55 : SUCCESS in 14m 24s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4cf211a9d9394a5c83ecd813cd11c84a : SUCCESS in 1h 57m 50s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8ccc09234e54720aa95c855a391eceb : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc699859499e412b9cedc2e3aef8039c : SUCCESS in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/93136cffa684417da071464e2806281c : FAILURE in 16m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e88c45e396e4267892e24a91fefb16b : FAILURE in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e5e7472854f754fb56a6c5fb471312a1692d6e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:27:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82cd0ef80dd745b0b5056c7242265966\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a2f26dcfdb864285831ef1f386eda417 : SUCCESS in 17m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bab7b873ed2a4ca5ab4c691284929bda : SUCCESS in 1h 02m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/04e72bbc51ed4d0ea4a75f666ce37e5b : SUCCESS in 1h 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4e2e727bd749a2a730a6019c487ba4 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9bc9dd1caf14353b3a7842873a526c3 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9a9ba8708f2443a94dbb1025fbe5e91 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b52c00b2677448e9730fd7111fd8ba7 : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1926ff00b2a14c1fada9c43d673244bb : SUCCESS in 18m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b50e67f703524c0c87147602135ae810 : SUCCESS in 1h 06m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5870b136aa1041a28e205661d7148a2e : SUCCESS in 1h 26m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5b55a8e7056d49b4889bc84d4fd6cb83 : SUCCESS in 22m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4d7e326b7b404565ad48f8735200e10d : SUCCESS in 46m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/be1fede8470447e8a4f9158a7aa81edc : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":3},{"id":"170315d518aced0eee5b0eaabef9ce52b806a472","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4e22366f88286ef92246551b21f6f535ea326bd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:28:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a86f77dba3f4d62b7bf8b8bf3663819\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cc32ad680c4f5c8c97fcdc5544fec1 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7d43bed84dff0c394a28730fef593bdfbb064f2","revisions":{"c276d51e3d202e12e8c92f79802e0e892d40ea4e":{"kind":"REWORK","_number":1,"created":"2023-01-31 10:28:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/872270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/872270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/872270/1"}}},"commit":{"parents":[{"commit":"0c8ce111baaaee360c02b569f6dec14307471f9d","subject":"Merge \"use underscore for unused variable in loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8ce111baaaee360c02b569f6dec14307471f9d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 10:27:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 10:28:34.000000000","tz":60},"subject":"neutron: Add iproute-tc","message":"neutron: Add iproute-tc\n\nCloses-Bug: #2000782\nChange-Id: I5cf944c509f23d430665de458cbbdb2fd31db7bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c276d51e3d202e12e8c92f79802e0e892d40ea4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c276d51e3d202e12e8c92f79802e0e892d40ea4e"}]},"branch":"refs/heads/master"},"3f866c775380a3aae392d804761ec358dd44876c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-17 08:19:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/872270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/872270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/872270/2"}}},"commit":{"parents":[{"commit":"12d431e3994023ce735a2352554a99a5cc9431c9","subject":"https: remove dhparams.pem before running gencerts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12d431e3994023ce735a2352554a99a5cc9431c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 10:27:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 08:19:37.000000000","tz":0},"subject":"neutron: Add iproute-tc","message":"neutron: Add iproute-tc\n\nCloses-Bug: #2000782\nChange-Id: I5cf944c509f23d430665de458cbbdb2fd31db7bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f866c775380a3aae392d804761ec358dd44876c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f866c775380a3aae392d804761ec358dd44876c"}]},"branch":"refs/heads/master","description":"Rebase"},"d7d43bed84dff0c394a28730fef593bdfbb064f2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-22 10:59:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/872270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/872270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/872270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/872270/3"}}},"commit":{"parents":[{"commit":"bcfd77bae7a6bc4c8a4aa5fcbee55f121022694c","subject":"Merge \"base: make curl retry on all errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcfd77bae7a6bc4c8a4aa5fcbee55f121022694c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 10:27:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-22 10:59:34.000000000","tz":0},"subject":"neutron: Add iproute-tc","message":"neutron: Add iproute-tc\n\nCloses-Bug: #2000782\nChange-Id: I5cf944c509f23d430665de458cbbdb2fd31db7bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d43bed84dff0c394a28730fef593bdfbb064f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d43bed84dff0c394a28730fef593bdfbb064f2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
