)]}'
{"id":"openstack%2Fkolla~872325","triplet_id":"openstack%2Fkolla~master~I0832d7997536f5f0dc195d5038545b6cc220f0fa","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-08 11:53:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0832d7997536f5f0dc195d5038545b6cc220f0fa","subject":"CI: store only templated out Dockerfiles in work_dir","status":"MERGED","created":"2023-01-31 15:50:27.000000000","updated":"2023-02-08 11:54:26.000000000","submitted":"2023-02-08 11:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872325","meta_rev_id":"406362092440d040b4e08a1c9e28cde42768198e","_number":872325,"virtual_id_number":872325,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 11:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-08 09:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-08 09:50:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-31 17:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 09:50:19.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9bc824b3d9d80ee81b5a0a3eec2ae21c296b273f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 15:50:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bceb44972fd8f586116a41f7f576fc6b6a2496d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 16:01:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95bed53a2461fb0a665a588aa1ed530c330e3505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:36:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe1cc70357414476956ff96d7f3c7d00\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/db9811419c404529ba7ae2a2ac54e052 : SUCCESS in 18m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f4516887e1b04fe49408d740672a694a : SUCCESS in 1h 24m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1686c0a504184f438375bcb908d45a74 : FAILURE in 20m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c78f16f7ffff4715aced456b7f5bce54 : SUCCESS in 45m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a3590d2fc0bd4906879a6551c54451ec : SUCCESS in 56m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a18cc469a5b4b20a9544bb3ea073b25 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1de8164171a7477c8932e5252cff8710 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecce6eb0a97a4b67845d14b3e89f027d : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/379a78727c764a3884aca68946c3ae45 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59adfe9d83314575802d478fcf552625 : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f1257ef78f244ed49129fd5771132f4f : SUCCESS in 23m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0e76cda3eb8342699ff559a850b5b6bc : SUCCESS in 1h 19m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8655ab0767284d6aa92d64629649cb8c : FAILURE in 26m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/58d4f62aca41488daee281639a92097a : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/77bf079e09894374889ea5e8e37f1710 : SUCCESS in 1h 13m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e75e7ab262c4c585e18f2dfb1016cf46157ffe5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 19:07:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e84dffb000e4bc699255da312194fa5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/213a7c85371a4977a7b9bde19d755e79 : SUCCESS in 2h 51m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7a4e89ffea2c44afa5b336237fb0b4c9 : TIMED_OUT in 3h 03m 03s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c96b63590e64ddb8257c64cc08eae8e : SUCCESS in 2h 58m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd803a3c0ab844d39fba203338f83797 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4fa86831376452d928c2e2a29b0fc4a : SUCCESS in 3m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b45a40e7aadb4a9cafedf5091388f6fd : FAILURE in 2h 20m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/49aa15af5ed443499eba6daac0f37654 : FAILURE in 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0107091dc1024f7ad990dca04b9d9479d7680164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-07 14:23:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ed1a5e1ad329629c74669658398d799a1dcd8007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 16:19:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d633a989b44f4f97de7eb13149c57c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e722b668fb0446108f02be52aff467d5 : SUCCESS in 30m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f904343b06ab4687a081f08f4a010d07 : SUCCESS in 1h 21m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e5c018073ee44363bf76577f0d3ec03d : SUCCESS in 17m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5c0e36c0d47b4733a0556e9a42c3c95c : SUCCESS in 59m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/05c2fb8e79f14a97ac292862024851a2 : SUCCESS in 1h 07m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceafe16dfe64465b90049c9f221d8aa1 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e806eb6d8f64c23ad02ca2c90fc8180 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47cae8e75e554f3daa49035cb4e3e999 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce0200e3e2048f58612984dec6ecd1e : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474746f9530246ee9823a32a4dfa4478 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f549d08344ee43e8b2e01e9d0d4b1fb5 : SUCCESS in 19m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d7ec1109356a43a5a32e60bce66f39c1 : SUCCESS in 1h 13m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/52df6f13d4a24b6ba4d9567980902fce : SUCCESS in 19m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7ffd38137aa742adb6f490a6937f4546 : SUCCESS in 1h 08m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cbe0f5fe1e2a4470859a5842af2c468d : SUCCESS in 1h 48m 41s","accounts_in_message":[],"_revision_number":3},{"id":"cb2de3335636b484c03e13a4437ac6956338f32c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 18:46:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4199fcb3884d38b6a3903aa4187368\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ccc8dc8b68e847cabfc2e3c6c3a44359 : SUCCESS in 1h 33m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/28f9c0fda24a44048f1aa45ecd215b28 : SUCCESS in 30m 18s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/37bb805f692549d2885dac9a9ad2d913 : SUCCESS in 2h 15m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8ed95ec2def4f519aa296d59c0cc8c7 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/550994ca3bbb4c19b55501afce2400f8 : SUCCESS in 6m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2577c0ae556f400183f0674c19e237a9 : FAILURE in 34m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d3fe340c669247229fb47884ef14422f : SUCCESS in 1h 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc177d40e04a14b9b6cd7dcc2149b71bf0478ce2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-08 09:50:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"443aa72bb3d8bba604f4799bae8a63937e41769c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:51:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5378f79ee187d27cc6b688f0c0801bd956027ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:53:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03ad2af192b94ddebd7ea39a691c0c91\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25823d0503844a3eb5ddee0799b3bddc : SUCCESS in 17m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dd2ee9aa42f04ba8b971c808b2390c53 : SUCCESS in 1h 01m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a4b41c20e9d84872a826209877346a3d : SUCCESS in 1h 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33361411029741c683597a974f1cbed3 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe958ce3e021483aa25ab99286e966a6 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/980651b5de614e90a4e0f1930b2ff2d2 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e0d59f333247b2a5974306f1e7aa1a : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/548319a096dd4fb69306670079675a9e : SUCCESS in 20m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6e5f88f6af5b475d829c0a6f25d01e31 : SUCCESS in 1h 14m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f8f0d710f3304433a240b6a76ba918bb : SUCCESS in 33m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6ea2f4dca68946728e310ef6752eb4e8 : SUCCESS in 57m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1e77b2d93574ebfbfa95930c99aebf6 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":3},{"id":"5e3c1effe7377115025d135054ca5543cb6e95de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:53:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"406362092440d040b4e08a1c9e28cde42768198e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:54:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50641c5b322f43c19e7897534e6cbead\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307fd999e4274bbf8b62f9e898d7789a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c45a742e6fbfd3392511b315edde17ab42a9760d","revisions":{"5e5437cc6aa385ba99e43e80dcf08c0a9fbd530a":{"kind":"REWORK","_number":1,"created":"2023-01-31 15:50:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/872325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/872325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/872325/1"}}},"commit":{"parents":[{"commit":"0c8ce111baaaee360c02b569f6dec14307471f9d","subject":"Merge \"use underscore for unused variable in loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8ce111baaaee360c02b569f6dec14307471f9d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:49:53.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:50:24.000000000","tz":60},"subject":"CI: store only templated out Dockerfiles in work_dir","message":"CI: store only templated out Dockerfiles in work_dir\n\nChange-Id: I0832d7997536f5f0dc195d5038545b6cc220f0fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e5437cc6aa385ba99e43e80dcf08c0a9fbd530a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e5437cc6aa385ba99e43e80dcf08c0a9fbd530a"}]},"branch":"refs/heads/master"},"748159dcbab752f4630a25e03c7e595c6ec8d0de":{"kind":"REWORK","_number":2,"created":"2023-01-31 16:01:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/872325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/872325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/872325/2"}}},"commit":{"parents":[{"commit":"0c8ce111baaaee360c02b569f6dec14307471f9d","subject":"Merge \"use underscore for unused variable in loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8ce111baaaee360c02b569f6dec14307471f9d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:49:53.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 16:01:16.000000000","tz":60},"subject":"CI: store only templated out Dockerfiles in work_dir","message":"CI: store only templated out Dockerfiles in work_dir\n\nChange-Id: I0832d7997536f5f0dc195d5038545b6cc220f0fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/748159dcbab752f4630a25e03c7e595c6ec8d0de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/748159dcbab752f4630a25e03c7e595c6ec8d0de"}]},"branch":"refs/heads/master"},"c45a742e6fbfd3392511b315edde17ab42a9760d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-07 14:23:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/872325/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/872325/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/872325/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/872325/3"}}},"commit":{"parents":[{"commit":"16f27c549660c4c27e3f440578ea1cf05b772a90","subject":"CI: Install Apparmor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16f27c549660c4c27e3f440578ea1cf05b772a90"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:49:53.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-07 14:23:32.000000000","tz":0},"subject":"CI: store only templated out Dockerfiles in work_dir","message":"CI: store only templated out Dockerfiles in work_dir\n\nChange-Id: I0832d7997536f5f0dc195d5038545b6cc220f0fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c45a742e6fbfd3392511b315edde17ab42a9760d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c45a742e6fbfd3392511b315edde17ab42a9760d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
