)]}'
{"id":"openstack%2Fkolla~872393","triplet_id":"openstack%2Fkolla~master~I4836806c0434caaff1e6fd1646aa5566b0c3d7e1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-08 11:25:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4836806c0434caaff1e6fd1646aa5566b0c3d7e1","subject":"rabbitmq: upgrade to 3.11","status":"MERGED","created":"2023-02-01 10:16:59.000000000","updated":"2023-02-08 11:26:44.000000000","submitted":"2023-02-08 11:25:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872393","meta_rev_id":"e6ab8aefa7a1ebb84b2151a27dab56916587c27e","_number":872393,"virtual_id_number":872393,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 11:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2023-02-07 14:09:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-02-08 09:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-02-08 09:49:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 11:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 14:09:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-02-08 09:49:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f9842966e29b40a58d071f6e7652cc84bdfdb7ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 10:16:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fa605fac95f37c3ef852c9c88bac45f4b151bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 10:26:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0063a6f67dd13581f26b6d3dd0649710a85c257d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:45:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9455df6136154e9a81546203973db273\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f458ac2947bb4d2ab73e72fbd5c2d07c : SUCCESS in 42m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/49406e6a1afa47afa3d2eba07627311b : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b43707c60664f9eaa937b4c96acb2cb : SUCCESS in 20m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/909a1e8c0e0b4c6c942228dcb8b286af : SUCCESS in 1h 00m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/97556d0f102c4bd4a63e332b07e39e3a : SUCCESS in 1h 09m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff4ce7ab7ce47ed9dc870932b8b901b : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d73ae8fdb0348f8a5ab8945b2a35c17 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/070d69558e0e4e06a4904c36fe5d2071 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e87bd4bfcede40499629d5cbf47e1b07 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560d36c183194044bb440eca87e76ab8 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c6dab00fa9046a69942df2d09022853 : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ba1a82ec417a455c92ecbb188393aad7 : SUCCESS in 29m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/91a2e891ab7b463db52d7925c5ca0eb3 : SUCCESS in 1h 11m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30a8308e11f04ef3a8ca5d5177bd7d3c : SUCCESS in 25m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1408c39fb1e44996878317a8abd5275d : SUCCESS in 46m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e03ddc7706b0499c9948cf1e3898c81f : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ccad302f86f3023a1387c89ea7fba7f258ae021b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:26:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ecfb6e0df1435293692b8344ac79ae\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b9898310c7ef445092f17776f1ebb50a : SUCCESS in 38m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d92188ae62c744aaac379ccfc7cd11fd : SUCCESS in 1h 37m 11s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b9ab906384d64ff4a2f2f3afc1f37cbe : SUCCESS in 1h 15m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dc9509f85c74a1aa59c246eece34852 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8dcb82452fbf4be7b608b7c17a75bff9 : SUCCESS in 5m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f6ce3ec6c11e4f9c95abc54d85adef89 : FAILURE in 1h 56m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5d8068ca14424a09a14dcb049c34f0a4 : SUCCESS in 1h 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa78563e4ce691a4ab965815e98270ac9284e64b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-07 14:09:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"efb46e00214d334dcdb4e02c7b6fb285f21552d6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-08 09:49:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8006c479bf30953103e07f3a1c4fd2ee7bf95ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:49:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60943ba654da2e4ed93962b00dc6702f2a81efca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:25:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4dffc779b7e4f188702909df1b4c7b3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5e56f8d920f441f89f76894236cd9e00 : SUCCESS in 38m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ad74738e6e2f45f2b74168136202ee9c : SUCCESS in 1h 05m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4c9a4d738f294f02af00289fa1185699 : SUCCESS in 1h 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b41a399cc244cdfb94f1013feef9550 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9669aba08d4244ee9b5b0ec70be0258b : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0db470c5452c434181e213aad48f48d1 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251d9d88d4b64715906d0ab64a42d106 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18031ad3045042f180f200ed95faab0a : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4aa692de9aa0498084a207665c3545cc : SUCCESS in 28m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ecb09cfa5e67406cb33546742b2bdd3e : SUCCESS in 1h 09m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a19474d4fd3742569546b6801cbad825 : SUCCESS in 39m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d422b52b21494446981945604c5b9bd9 : SUCCESS in 49m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cc18fbd603c14798b6ba3d71cbd49d6b : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0484c3ec55ebf394af3985cfdb13f3c591b70090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:25:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e6ab8aefa7a1ebb84b2151a27dab56916587c27e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:26:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb1de24425e46489403eeb22e6932d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503398b0fafb48a39964e28231ebb18d : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d1a63948f342f8874c4f9b22c76973 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9f4c78e87c8567f10900614d893a6d03d6b0584","revisions":{"b604560b6e165f0cb93e7ca0c6341de7bd2b51cb":{"kind":"REWORK","_number":1,"created":"2023-02-01 10:16:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/872393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/872393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/872393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/872393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/872393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/872393/1"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:16:47.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:16:57.000000000","tz":60},"subject":"rabbitmq: upgrade to 3.11","message":"rabbitmq: upgrade to 3.11\n\nChange-Id: I4836806c0434caaff1e6fd1646aa5566b0c3d7e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b604560b6e165f0cb93e7ca0c6341de7bd2b51cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b604560b6e165f0cb93e7ca0c6341de7bd2b51cb"}]},"branch":"refs/heads/master"},"d9f4c78e87c8567f10900614d893a6d03d6b0584":{"kind":"REWORK","_number":2,"created":"2023-02-01 10:26:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/872393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/872393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/872393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/872393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/872393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/872393/2"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:16:47.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:26:44.000000000","tz":60},"subject":"rabbitmq: upgrade to 3.11","message":"rabbitmq: upgrade to 3.11\n\nChange-Id: I4836806c0434caaff1e6fd1646aa5566b0c3d7e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9f4c78e87c8567f10900614d893a6d03d6b0584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9f4c78e87c8567f10900614d893a6d03d6b0584"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
