)]}'
{"id":"openstack%2Fkolla~872395","triplet_id":"openstack%2Fkolla~master~I17045980b6f73a0395b1bdfd3b1b6b63679284c9","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-08 01:34:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17045980b6f73a0395b1bdfd3b1b6b63679284c9","subject":"toolbox: bump ansible-core to 2.14","status":"MERGED","created":"2023-02-01 10:26:15.000000000","updated":"2023-02-08 01:36:24.000000000","submitted":"2023-02-08 01:34:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872395","meta_rev_id":"80024db5037479427c5d7f12d28d7aeea370321a","_number":872395,"virtual_id_number":872395,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 01:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2023-02-07 14:09:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-02-08 00:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-02-08 00:10:26.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 11:41:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 14:09:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-02-08 00:10:26.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"0b3a3f8b967acb78fcb1e190c674b849d7281b88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 10:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96217958581877ce2dce9ab342d8a7af136071b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:41:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b22bca6b663b4b3fb676086934c355e9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0f66282b86e446b2a4a43b586b948e2a : SUCCESS in 41m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/641fb8c668c74476b1d348a72ab9399a : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/93f08b0b553449159c4f192bee39b01e : SUCCESS in 33m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8197e349d63b47638f967c45880dfaf8 : SUCCESS in 1h 04m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/11c569fca387444fb956df17f502da44 : SUCCESS in 1h 05m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff88d2eda464a42bfdb2c691fe085fb : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/346fe70cfcbc407aa91e436870526b6f : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/640c48fbd0b441f2b4ce126bf8ca3c54 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcabb9e15f04449f9f684d77bcce74a1 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d05682e895a48558cc0b72a7b1443f6 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b352939eaf0146c3a199324b47c889e4 : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f956893e4ebc492cbe5b88a1cd078e98 : SUCCESS in 28m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/05de0b308ef24f91a362d6c902d15389 : SUCCESS in 1h 03m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82457578fe9f426ab5e7e137ca89d8be : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9432b7025add49b59ccf43a799a6dd98 : SUCCESS in 1h 03m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00b08e84082f4c3f8f429d47f8118700 : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e3336afd926f24ac48ef8077a8dc117e8b74cbcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:05:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4728ecc45d574fe28bf493609db97af3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d3bc9d6aecb84caf9b796ee7d9b29db6 : SUCCESS in 35m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e466ea2c985472688b6c923bc8bc6a5 : SUCCESS in 38m 57s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/44ce743c04654a0c88a0e896667c15af : SUCCESS in 1h 19m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d78684bda9aa49bca298349ea59a503f : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2efc3a0dca544e9e8dfa012014bb0170 : SUCCESS in 4m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/72fbb69e944443ccb018d844149a6757 : FAILURE in 1h 36m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/37d1cbb31d844075b8abca54e0ec968e : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bbc66898b26f81a45b44f8dccf672444cee1e60","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-07 14:09:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a7edb1c0f18e4e3a4be583f940d40ec63cf3d4fc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-02-08 00:10:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80ccec5294e51a44334df808f69f0a218f2d941a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 00:10:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b40894a3da3b9e63005bd28c1e455f71e79b2a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:34:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96e62835733d4d9690c48dff4e1d6a37\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d3043c0f0a3b457ca622abe9b00338a9 : SUCCESS in 22m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a6bb189472b84c0e9050bf81b7e62a6c : SUCCESS in 1h 12m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/36239db1e44046c7bbc675fa40a91f34 : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05d323239864df28f90beae36333eea : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7155cacf841a456f9d996bdf6d5cc599 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f634b0d263c4b05bfd1d54f3adfdbe4 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf65043b1be049d284610cf01e0103d8 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb90b17c53594476a83774032e1acc08 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/38316e312be64e8796eee301dd2dc176 : SUCCESS in 40m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c7702c7e2179423c8154d52a70ab8921 : SUCCESS in 1h 06m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/793a0c0934e14e45a1c16910f46214a3 : SUCCESS in 21m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/81b89c0736fd4e72811e720380091ef0 : SUCCESS in 47m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bad315cf2b194936a0e23748138854e6 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"db3e24dc8af669509ccc9504216fb7cc056fda83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:34:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80024db5037479427c5d7f12d28d7aeea370321a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:36:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20e660b04f64d8eae74c5a7be01dfb4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f76b2e9f2e463d9b5807e42f90be15 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3e8e44367541c3bd295826cbad039f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61f649febef74d46baec4dae546ba7fd0e050762","revisions":{"61f649febef74d46baec4dae546ba7fd0e050762":{"kind":"REWORK","_number":1,"created":"2023-02-01 10:26:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/95/872395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/872395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/872395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/872395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/872395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/872395/1"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:25:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:26:13.000000000","tz":60},"subject":"toolbox: bump ansible-core to 2.14","message":"toolbox: bump ansible-core to 2.14\n\nChange-Id: I17045980b6f73a0395b1bdfd3b1b6b63679284c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f649febef74d46baec4dae546ba7fd0e050762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f649febef74d46baec4dae546ba7fd0e050762"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
