)]}'
{"id":"openstack%2Fkolla~872488","triplet_id":"openstack%2Fkolla~master~I6c6ed9f1739221666619242fc759cb2e464f7390","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-02-02 12:50:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-05 07:22:42.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-02-05 07:22:42.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-02-02 08:43:49.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I6c6ed9f1739221666619242fc759cb2e464f7390","subject":"Fix wrong source names for skyline in kolla-build-config roles","status":"ABANDONED","created":"2023-02-02 03:40:53.000000000","updated":"2023-02-05 07:22:42.000000000","total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"7ee2bbbec84b148b704b5f2d4511c0f14543ec33","_number":872488,"virtual_id_number":872488,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-02 05:07:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2023-02-02 10:52:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-02 08:43:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 03:41:21.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-02-02 03:42:18.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-02-02 05:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 06:26:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6a3615caeb501e4234f8518a1345d6aa0fef1032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-02 03:40:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00d4769493bdd0a08db30dba2351d9da4931982d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-02 03:44:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cc1aa332a83df40968682c49efb36d0b4fea2d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 05:07:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eacbc4e067f3476e8313ff12f4cf555c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/38d6bc067e58436c9a17b85ae99093e8 : SUCCESS in 43m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1fa2d2aaad5744fb9180ed1f3f30827d : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7497016d40fc426faa9ac5bb0ed8bace : SUCCESS in 34m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/863b20ea671d4d00add2412a66762d0a : SUCCESS in 1h 07m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c944581415e949e8a53c4302fb269596 : SUCCESS in 1h 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/356ef04b3a4d4db883ee9d9ed6b03a6e : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8af8890d24c540288adc026cd57bc4f6 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcabaed5c6e04cd79608f49d9b12eba9 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59d6c86032ab4548b732147a28b98106 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e994f45090fc40ee99f7fe3cbe0a2f3a : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc7a1708009344f58c866114104ad9b9 : SUCCESS in 26m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/836e653254ea4987bc5dbcc46ec77951 : SUCCESS in 1h 01m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc96b539350c4a2d8eaf6e81e557df04 : SUCCESS in 26m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ab851435e77b4d3590151023b025bca3 : SUCCESS in 48m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/97e400bce2a742e8aeeffb7c2af8a9b5 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c98b5176946777b0956d7a8637f1f7347df20fcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 06:18:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d095794825ee4358b58163fc85a3c2de\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6dca873e95394354b17eb4ff2bb96714 : SUCCESS in 1h 58m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e12331bbdacc45759caf94b94321469f : SUCCESS in 2h 12m 30s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d72d8637f0484e32b618103dc51d87cf : SUCCESS in 2h 30m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c03fda864a4c48ec8e12ae2dd4b297bf : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b925223739ca446d81be35443461f5f5 : SUCCESS in 4m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/24cb38f6bcf7406ba1085701b472b105 : FAILURE in 2h 05m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ef3665b3526144798ce8a6efb384e9be : SUCCESS in 2h 29m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ca4d9d044934a11f5b7cd87da0bd82633381b34","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 06:26:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1c0aa73484b004afe81f3d2c674df98575bf45dd","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-02 08:43:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d197988ab27f50ac06fc46ac21b8a7d7dc93900","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 10:50:37.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f66ace469ac90f20fb590ab8605d96d230fbeca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 10:52:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf28ddc58b5ceaa3513f5c324ce8451e078003f2","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-02 12:50:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ee2bbbec84b148b704b5f2d4511c0f14543ec33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-02-05 07:22:42.000000000","message":"Abandoned\n\nNo use","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"970455e0f2d2bf91195a155834adcb5cf18762ce","revisions":{"ee88a4d01798bd5a933a348525a7801f34e89e70":{"kind":"REWORK","_number":1,"created":"2023-02-02 03:40:53.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/88/872488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/872488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/872488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/872488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/872488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/872488/1"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-02-02 03:39:01.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-02-02 03:39:01.000000000","tz":480},"subject":"Fix wrong source names for skyline in kolla-build-config roles","message":"Fix wrong source names for skyline in kolla-build-config roles\n\nWe use skyline-apiserver and skyline-console as source for skyline.\nSo we remove skyline-base, and add skyline-apiserver and\nskyline-console.\n\nChange-Id: I6c6ed9f1739221666619242fc759cb2e464f7390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee88a4d01798bd5a933a348525a7801f34e89e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee88a4d01798bd5a933a348525a7801f34e89e70"}]},"branch":"refs/heads/master"},"970455e0f2d2bf91195a155834adcb5cf18762ce":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-02 03:44:57.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/88/872488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/872488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/872488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/872488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/872488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/872488/2"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-02-02 03:39:01.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-02-02 03:42:35.000000000","tz":480},"subject":"Fix wrong source names for skyline in kolla-build-config roles","message":"Fix wrong source names for skyline in kolla-build-config roles\n\nWe use skyline-apiserver and skyline-console as source for skyline.\nSo we remove skyline-base, and add skyline-apiserver and\nskyline-console.\nSo that the periodic task will build and publish the skyline images.\n\nChange-Id: I6c6ed9f1739221666619242fc759cb2e464f7390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/970455e0f2d2bf91195a155834adcb5cf18762ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/970455e0f2d2bf91195a155834adcb5cf18762ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
