)]}'
{"id":"openstack%2Fkolla~872526","triplet_id":"openstack%2Fkolla~master~I4b0ef522ed0812801a6a1968981b833debe38df1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-22 10:57:20.000000000","reason":"Change was abandoned"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-03-22 10:57:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4b0ef522ed0812801a6a1968981b833debe38df1","subject":"CI: Use CentOS wheel on Rocky9/aarch64","status":"ABANDONED","created":"2023-02-02 10:59:48.000000000","updated":"2023-03-22 10:57:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15c16265f9471a28335f16ca2ed0ed5b95c4043b","_number":872526,"virtual_id_number":872526,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2023-03-17 12:27:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-02-27 02:27:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-02-22 13:30:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-03-17 10:59:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-03-17 10:59:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 12:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 14:35:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-02-22 13:30:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-02-27 02:27:38.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-03-17 10:59:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"d8b0384034c710f209d41b2ba38ad83423579165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 10:59:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fcc7dbd563f76e2df2afd305e76a2674583639c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 12:05:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9be3c06a297a43c29feb8f216fe2d970\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/06a961f200de49139d7c1a8722e31fa8 : SUCCESS in 28m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/560288902a7e44df8d8088e56537605c : FAILURE in 50m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/58e8a3eb7eb44f1dbbf59b12555db759 : FAILURE in 3m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23ff877373a048948bf3a8494fdadc9b : FAILURE in 15m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1c29f6cfd13446b9bd30b735c975d64f : FAILURE in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3118186439bb4e439db3cf7020fcdc36 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe873b15e2d4c80b6b28b1aeb5afe97 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ff3d7390e548f6b6dc540ee29fac2b : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4769017efa244fb97291ced07ef5a75 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425398fe495a4f9190935df5de7577d2 : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/664b81a000a04f58863442fbbb96d92b : SUCCESS in 37m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/36bd3aef2a484be78b9e1585e6aa0f35 : FAILURE in 53m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f04ea6691abd4523b3c33707b26ebf5b : FAILURE in 3m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4115d260c6c24f7d800d0d639c062e94 : FAILURE in 7m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/745fc4c9d2f04823b32c79d62b4e951c : FAILURE in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5e58346f851d5317baf66d575062c9d59e42af7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 12:21:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d3aa757e3f4b0083209ec4803f4067\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f9caad13fd074f34bdb85807f0bc495a : SUCCESS in 23m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79c107b130754774805d406fe75fe5d0 : FAILURE in 3m 49s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02b9d63f3b204202a21605ea19edf421 : FAILURE in 17m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4150c7352fd241efb04f94843022a687 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34e71c3cfae7437b9d2af9fbd385020e : SUCCESS in 3m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/452518ca105a45dd9639334536958387 : SUCCESS in 1h 17m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/720dc22286de4b4e94f55a4efc649bc3 : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5319acce4de603b0bd1a663a116c322606182561","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-07 13:49:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e3d5ad5147279765cc87bd2a3319e9d625951c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:17:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/371d3e52936841adbfa88e19ba64a01b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/af5602118ae2417db607623b8992defd : SUCCESS in 26m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/76c5940b590c470aaced04d5062d500b : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b9a515ddd8941fe8a660144e737f955 : FAILURE in 3m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dc3da641c63b4554b84b25605147332e : SUCCESS in 1h 07m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ce4515bda96642e49935bfb9dedcf2d2 : SUCCESS in 1h 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013f8f92ab144825bca71a1efe81592d : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09455e6932ac41eeb6aa8f90e69e2770 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25ad197844464a128df19781dbdbf46c : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53722ee7d354caebb43f012fd0ff239 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad63b820f6454f9db5aca238442abcdd : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6de6c03903644a409f89b37cd3117d93 : SUCCESS in 28m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/47b936b78691441aa28f1551758ccff9 : SUCCESS in 1h 09m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8515ef1007b5478ebafe254415a3be55 : FAILURE in 3m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/168879006ae3458ebde97e37dad415a4 : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c167041d49ed4a459e8708fa5447c5f9 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"460fb97ec66eb805252f7c7c409f56659adf3431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:23:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15144e67c9f540aea96c1f7519f490e5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b9b50561f14d4bbfb05e7a050d575a56 : SUCCESS in 31m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e20e0fa971d9472794b2a3c04d4484b0 : FAILURE in 3m 55s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a42eee2f1e8b43fcb591b9832b792878 : SUCCESS in 1h 31m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ffc7170103a48a4809275c0a1aacc66 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/066482bd36d3483298543798a2cef474 : SUCCESS in 3m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eb287d88734440b4afbefdad54c52f3c : SUCCESS in 30m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f349d996b38d4a33a6ed84568bf46b0b : FAILURE in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b9796c2bbdc0faa06e94b5edea9cf98cdb8c0be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-08 09:50:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"66e7cedbfe91e7d4006664b4e8ee9ca2acda7315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:26:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c268f7e05bc54105aba80725f634d279\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/accb43367344472aabd810c0ea1ed0ae : SUCCESS in 43m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d68245d3f50e406e8cbecf071038b531 : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41794995949740698dab1a5f7124555e : SUCCESS in 39m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5b36f6ec906b4938b5a529574b17e3b7 : SUCCESS in 51m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3ea35ab219f64b9e906db90700d3b985 : SUCCESS in 1h 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68c620b90f54eb7a742518ddd1a3dc8 : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b69a1c88c016477a8e8479ef145f722a : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4928dbb5aa124d55a759cf706dcfbdd7 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c3140657aaa45f6a009c0e575e5da6f : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bc66c8c4004f9a98f36dce9447ad49 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/094219d7e89f44be8f9073744f311051 : SUCCESS in 28m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0c88cadb8eee4871abfe3614854f6934 : SUCCESS in 1h 04m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e819039bb5454073946484b896f14671 : SUCCESS in 27m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0b112760e0094bae87a416d5c8f234f2 : SUCCESS in 44m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dcd7bd51eeee4121a949bce65a41fca7 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":2},{"id":"584ec7a888d8cac9df71946ca274967781c23f95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:29:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e907d256c43f477e8eb17a29a8029a70\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/70c2d333daed45f8ad28b204c73b2b1e : SUCCESS in 31m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8298cf686ce54655a334f059a22bdcaa : SUCCESS in 29m 43s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/89355f6f3c26477a8981627e90bc7a39 : SUCCESS in 1h 27m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae6206c603e4446d9a1e790ab8263d97 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8939dc827c854d559c9d03ae7294d3b8 : SUCCESS in 2m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d0eb678997eb43beab571a51955583dd : SUCCESS in 1h 35m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1ca39117353c4017ada8f765eacc7fb2 : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fd5ca24bcbf38b28cba329fbbed8f1d5aea704b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-22 13:30:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"531e119188f946c85aad2c2091d003dbb2682ea5","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:27:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3918c6cfbaf419f226a7987ebef08467014292c9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-17 10:59:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0d37c11f0b172f66360fd437f13a56f5c51adbe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 10:59:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78382208ace9e045ea22f90271918506d4c16644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 12:27:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/510749601eb741098089c123aeb2f981\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7bbc738120324dc0b072155f620cf11c : SUCCESS in 25m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5ecc717220294d1d97b5ce048523e7fa : SUCCESS in 55m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/75281b060cf3437f8e0ba0605221b178 : SUCCESS in 1h 07m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fedb995cc9634e58ab752461baf72520 : FAILURE in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f02cc63e96714029840a8267933cfebd : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18d9ee8008474d1389f8cf3af8dc2c5c : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9eec538e4b24c8caf9639b8358e8ceb : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d8dadfd502bd4e4ab46a2534ee14d156 : FAILURE in 40m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2255d877327b492994144ec55063526d : SUCCESS in 1h 18m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8629237b01a7460486259b457b20a19f : SUCCESS in 1h 23m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e3022fe3989f42508a8b97ec807053c8 : FAILURE in 25m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/02cd3acc4ebb4185be4934308ae2a85f : FAILURE in 15m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/30d0618949634bd7a9f0caa4b276d455 : FAILURE in 22m 00s","accounts_in_message":[],"_revision_number":2},{"id":"15c16265f9471a28335f16ca2ed0ed5b95c4043b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-22 10:57:20.000000000","message":"Abandoned\n\nbeing fixed in a different patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d8afd263b4c8b3522e5b918ba5891ba89491a050","revisions":{"2c4d94379bf0fb1a648e3976819c1aec40f2b633":{"kind":"REWORK","_number":1,"created":"2023-02-02 10:59:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/872526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/872526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/872526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/872526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/872526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/872526/1"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 10:59:23.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 10:59:46.000000000","tz":60},"subject":"CI: Use CentOS wheel on Rocky9/aarch64","message":"CI: Use CentOS wheel on Rocky9/aarch64\n\nChange-Id: I4b0ef522ed0812801a6a1968981b833debe38df1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4d94379bf0fb1a648e3976819c1aec40f2b633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4d94379bf0fb1a648e3976819c1aec40f2b633"}]},"branch":"refs/heads/master"},"d8afd263b4c8b3522e5b918ba5891ba89491a050":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-08 09:50:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/26/872526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/872526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/872526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/872526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/872526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/872526/2"}}},"commit":{"parents":[{"commit":"f3e5a90ca9d3185e8adbe9d60ff4aab19d1a2bf8","subject":"Merge \"toolbox: bump ansible-core to 2.14\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3e5a90ca9d3185e8adbe9d60ff4aab19d1a2bf8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 10:59:23.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-08 09:50:56.000000000","tz":0},"subject":"CI: Use CentOS wheel on Rocky9/aarch64","message":"CI: Use CentOS wheel on Rocky9/aarch64\n\nChange-Id: I4b0ef522ed0812801a6a1968981b833debe38df1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8afd263b4c8b3522e5b918ba5891ba89491a050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8afd263b4c8b3522e5b918ba5891ba89491a050"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
