)]}'
{"id":"openstack%2Fkolla~872557","triplet_id":"openstack%2Fkolla~master~I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-07 12:45:00.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-02-07 12:42:34.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d","subject":"CI: Install Apparmor","status":"MERGED","created":"2023-02-02 13:40:34.000000000","updated":"2023-02-07 14:20:10.000000000","submitted":"2023-02-07 14:12:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872557","meta_rev_id":"cc7be4dcc055ac02c819e3ae1dbad7192b804696","_number":872557,"virtual_id_number":872557,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 14:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-07 12:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-07 12:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-07 12:45:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 14:32:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-02 14:32:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 12:41:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"d9f71e520fe51c8d5e3e5dfd0b3e884e3c83ccfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 13:40:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5978d8c104cf72057570460f6ea615d81fe28076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 13:41:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e9b5a59506b3527b31d62972d9230369684e1a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 14:22:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd609c5718faf9234753bc3239ecfeed8dd040f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 14:24:15.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4feea3bad96c3134cbde8af1411fee99c48034fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:32:48.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872558 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"0b55a6ac2f9a7781dd9646e291fb8c4a1f101550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:32:50.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872558 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"8a5ae1c8439b193fcd264353609cfa7cf9f83f5c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 14:36:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c982fb5afe640d2173946e1c973bdd62868eb19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:54:50.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872558 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"2cc3ad7ce4e59d0889a25114ae01cdbdba992a2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 14:55:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"6a249ea36a10842ae00c1e6aa656e3d8addbeae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 16:27:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fbe7974c59b4fac900884d29ade52b0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/78979ac0a1774cc988734af44744a431 : SUCCESS in 26m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/47abc741f58c4f29a5be16f7266daf37 : FAILURE in 54m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6fbe166da0e54ce6a628feedb6e6d18c : SUCCESS in 42m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d864261f999845c29185bbacb2b5e4be : FAILURE in 43m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e4d8dfd93b334d61acfc6550bf866484 : FAILURE in 1h 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3d19786504466a96b34e9377e194f8 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733cb5f78e8f41ea869e776e85958a76 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1264d38af6284ed3b8194fdf2b261763 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/181ec56410324fefb5496410e0195f9f : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395780a7a7174855a72245f6458cc437 : SUCCESS in 4m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f510523aec154ec9b345d8c68faaaf34 : SUCCESS in 28m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/44b9984c69664e579dfefd7986188e9e : FAILURE in 53m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e2212d2c0bc4c738151d606c1cec7d6 : SUCCESS in 43m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/927948799bd34c95b37804f1b4299366 : FAILURE in 39m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a445ad497e1142deb867e0ad84c16e64 : FAILURE in 55m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ee85d067884426bcb138dd629d84c2400601ed79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 17:39:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1062e824f46648549bc1d8b95a3206d7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1799930c9236478ab8679a6e6a76abb1 : SUCCESS in 28m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1794587ad06a49d1a685c43a13105a16 : SUCCESS in 27m 39s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f80d8e68a44482596574d47553fe0dd : FAILURE in 50m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77c5db2d932a47e59010556dbb4c6a2c : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8aa74ccd82cb4f7e8a6f00dddb3baf3d : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e72f7b59f7c74f2695fb6d3743139436 : FAILURE in 2h 41m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c5e8e6bde5bc4572a0f3aa9436a9c796 : SUCCESS in 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a363344952658f9024f1f9f1fcfc9ad33eae0927","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-03 08:56:10.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"ee9740b785f114ffe251759c8f8b1a68aa978706","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 10:10:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4e1a3eb3f34fc983d34eae59870850\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8cf83a43e30440df8eb7cc261fc77b76 : SUCCESS in 33m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ae9281dcee254a108665c829b1cd7389 : SUCCESS in 35m 28s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/976e7a2f11e44708bd1626f66d9d6b3d : SUCCESS in 1h 11m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d18e8efd43e844e7aae9baf94c0bfe18 : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/945290a37280464a9514abbb0f570353 : SUCCESS in 7m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2766f3c6e9564158bf3ab7973cd11c2f : FAILURE in 31m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9a33acd9e49a4577a9860ea29612ab82 : SUCCESS in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"61cfa48911bb3f5ed788acfd591d70521a58fee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 10:14:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be4598c9c1ff48bf91474665fe59df2c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ac3f43048a154d2c8cee9c096b7663a7 : SUCCESS in 32m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8ea8d1844483447a90a42b9e4c63db2a : SUCCESS in 1h 11m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8807072c4325451a95a7adf6f4fc4588 : SUCCESS in 28m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/65b8a314683b48ebb8f47b98dd7e5664 : SUCCESS in 1h 03m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0127880088464f88ba5f479242220ec6 : FAILURE in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d749af929174427add0d1a05d1e0aea : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc94b746ddaf4d98bbcfffd452cef8b8 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29a0855bbaa147cc82e0258d15f706ff : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad09ff25219425d912c05dde2b6146d : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bb7fa42767481ea9c719a835ed4761 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3264baea91a7445c983d2234c410157c : SUCCESS in 24m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5b593d40031f4fc2873e092e6d557e02 : SUCCESS in 1h 05m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40fca11937774594863ea332640dc652 : SUCCESS in 29m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f0c640f8d9414ac19e7980f0d19e8a4a : SUCCESS in 46m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a2a93d9d470f4379a76420a307457bf4 : FAILURE in 9m 23s","accounts_in_message":[],"_revision_number":6},{"id":"eefb80c684b2aa093e5ce90c8647b72baf9f7214","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-07 10:53:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b1c8e88e5b2fe5c4a0fd9325a7972a25bd0957b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:12:24.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872626 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"f52fc62f37fdcccd64d9ee72ed4aea94309fd06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:12:35.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872626 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"3a580b9fa832d6353dd378f69acb496ec32b7a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-07 11:18:10.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"08a5d5e7929cef9b4cf9f43b397f31f957159917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 12:42:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae43966c76d44579b566f3848cd3e9c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4a472a70aae84f85a0dd42db8d8e9712 : SUCCESS in 35m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8f74de92fedc46aa90e24074a8866067 : SUCCESS in 1h 12m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8606514b220415e969e2519f906c6ab : SUCCESS in 32m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/227b34509d93433583ccb2821da7d34a : SUCCESS in 1h 10m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e682bf02a6e7490aa34f419f68f22f48 : SUCCESS in 1h 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1163e5f2ac1450b9f4888044e5233a3 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/516563bfb0e741b4b6253b3096cfbb92 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c5d9fa1099f414b91e7a3faa8a903b0 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1d66280758641bd8bf948d7a339af4b : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8f5656eb57424cabc3f73a69af7b45 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/50ef2f1b05cb4664ac3f547f4d2c31c9 : SUCCESS in 36m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e1ba88e1b1c497c827fcec5f48002b5 : SUCCESS in 1h 03m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9cec11f58e1446b2a728bd56c27539dd : SUCCESS in 30m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/655407da7ac44190be4e8f15cba2ae0f : SUCCESS in 1h 01m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b3bfe787eeb342e3a5e070358bbf6669 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":7},{"id":"ca394d62acc69932d5905676ec5d946314c6ce8e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-07 12:42:34.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d05d2a739afd37ba71ed95f9a32ad55382eeecc2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-07 12:45:00.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ba7db79c7ac5f4abd38f935f0a54caa235f337e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 12:45:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"36aff6aadb8e098364525de272e7bec07ca13e36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 13:48:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ca5ad5a2c9463a8ddc4d8c0696d29f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ae212b25669e402bbedc4898db2d230c : SUCCESS in 25m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dc79f04648b04fbb9c12927ff5758ddd : SUCCESS in 25m 29s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90e5137bd3064aadb8cc3ae55ed7aa0d : SUCCESS in 1h 11m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a662cd9ab6c4e2695f8ef27f0902d68 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/960598bbd1d042229d5875a43029af35 : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/906b6acec6ae4ac6b05587debc104209 : FAILURE in 2h 25m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/edb29721b31e417787533b9fab1c9e9d : SUCCESS in 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7489f1974af341d0f925f4f7d9c8b28cf28e2b5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 14:12:37.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45179bf911ef4788934704b53a2b810e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a6a7a2ca1c54dca9bcf06afc471905c : SUCCESS in 32m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bfe147b1843442c0a2962873adfc741c : SUCCESS in 1h 06m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/324b29bdea194f3bb846d14583cc7802 : SUCCESS in 1h 08m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ac3dc7f22d4d1caf84f508e115116b : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9b07a9a9af34e61b4c40e992d65aa8c : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70ea01354ab64f5ba2cca11cc2289911 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd6e12303914c70963b542846b484b0 : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/773668ba4c6748519e3576f3e922f74a : SUCCESS in 28m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8a77713155b64c04a7edb3e20af18c2c : SUCCESS in 1h 11m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cddf025dc29d4933a49f62d2ff879a71 : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/644ae15808f84817a6015a7740eae77c : SUCCESS in 54m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/732bf94588c246b4800608972d47de7f : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":7},{"id":"a2ae05791f915869b7960871138746b896b8d988","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 14:12:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"cc7be4dcc055ac02c819e3ae1dbad7192b804696","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 14:20:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e568e6aeeb458cae7f338782518992\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917d567e456e4d1689dc2fbfe1c9b6e4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"16f27c549660c4c27e3f440578ea1cf05b772a90","revisions":{"9fc335cded3ef167277742833ec2277a6ca0d353":{"kind":"REWORK","_number":1,"created":"2023-02-02 13:40:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/1"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:32.000000000","tz":60},"subject":"CI: pin docker-ce to 20.* on Debian/Ubuntu","message":"CI: pin docker-ce to 20.* on Debian/Ubuntu\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fc335cded3ef167277742833ec2277a6ca0d353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fc335cded3ef167277742833ec2277a6ca0d353"}]},"branch":"refs/heads/master"},"946fc8313f95c43799e1a665392be925a1bd8d45":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-02 13:41:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/2"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:41:20.000000000","tz":60},"subject":"CI: pin docker-ce to 20.* on Debian/Ubuntu","message":"CI: pin docker-ce to 20.* on Debian/Ubuntu\n\nDocker 23.* is out and it started failing with apparmor_parser\nrelated messages.\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/946fc8313f95c43799e1a665392be925a1bd8d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/946fc8313f95c43799e1a665392be925a1bd8d45"}]},"branch":"refs/heads/master"},"32201148aef958169dad2a1b74bb2983d1a1befe":{"kind":"REWORK","_number":3,"created":"2023-02-02 14:22:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/3"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 14:22:08.000000000","tz":60},"subject":"CI: Install Apparmor","message":"CI: Install Apparmor\n\nDocker 23.* is out and it started failing with apparmor_parser\nrelated messages - it seems apparmor is missing in the OpenDev image.\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32201148aef958169dad2a1b74bb2983d1a1befe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32201148aef958169dad2a1b74bb2983d1a1befe"}]},"branch":"refs/heads/master"},"5964c080c31235b2c04f0c795cc6ca6f5dd1e188":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-02-02 14:24:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/4"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 14:24:15.000000000","tz":0},"subject":"CI: Install Apparmor","message":"CI: Install Apparmor\n\nDocker 23.* is out and it started failing with apparmor_parser\nrelated messages - it seems apparmor is missing in the OpenDev image.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/872558\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5964c080c31235b2c04f0c795cc6ca6f5dd1e188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5964c080c31235b2c04f0c795cc6ca6f5dd1e188"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d8ada98856975a4224dad98632b338ad4895831b":{"kind":"REWORK","_number":5,"created":"2023-02-02 14:55:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/5"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 14:55:18.000000000","tz":0},"subject":"CI: Install Apparmor","message":"CI: Install Apparmor\n\nDocker 23.* is out and it started failing with apparmor_parser\nrelated messages - it seems apparmor is missing in the OpenDev image.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/872558\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8ada98856975a4224dad98632b338ad4895831b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8ada98856975a4224dad98632b338ad4895831b"}]},"branch":"refs/heads/master"},"25cd68628355e3919fd5f37b39a94f1e8125192d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-02-03 08:56:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/6"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-03 08:56:10.000000000","tz":0},"subject":"CI: Install Apparmor","message":"CI: Install Apparmor\n\nDocker 23.* is out and it started failing with apparmor_parser\nrelated messages - it seems apparmor is missing in the OpenDev image.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/872626\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25cd68628355e3919fd5f37b39a94f1e8125192d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25cd68628355e3919fd5f37b39a94f1e8125192d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"16f27c549660c4c27e3f440578ea1cf05b772a90":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-02-07 11:18:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/872557/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/872557/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/872557/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/872557/7"}}},"commit":{"parents":[{"commit":"5cc1f8c15e25bb1b62e8f36261ba4675578c5887","subject":"Merge \"skyline: Add new dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc1f8c15e25bb1b62e8f36261ba4675578c5887"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 13:40:18.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-07 11:18:10.000000000","tz":0},"subject":"CI: Install Apparmor","message":"CI: Install Apparmor\n\nDocker 23.* is out and it started failing with apparmor_parser\nrelated messages - it seems apparmor is missing in the OpenDev image.\n\nChange-Id: I2a6d98e3c1d2d1b7c97a8e6172decbf56e77042d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16f27c549660c4c27e3f440578ea1cf05b772a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16f27c549660c4c27e3f440578ea1cf05b772a90"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
