)]}'
{"id":"openstack%2Fkolla~873003","triplet_id":"openstack%2Fkolla~stable%2Fzed~I0832d7997536f5f0dc195d5038545b6cc220f0fa","project":"openstack/kolla","branch":"stable/zed","hashtags":[],"change_id":"I0832d7997536f5f0dc195d5038545b6cc220f0fa","subject":"CI: store only templated out Dockerfiles in work_dir","status":"MERGED","created":"2023-02-08 12:35:10.000000000","updated":"2023-02-08 15:43:54.000000000","submitted":"2023-02-08 15:42:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873003","cherry_pick_of_change":872325,"cherry_pick_of_patch_set":3,"meta_rev_id":"2031d925ebef69acf1235c16908ddbe32a61bf26","_number":873003,"virtual_id_number":873003,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 15:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-08 13:55:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-08 13:55:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-08 14:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e4b5ec1e2bcbb2b353412276b0adfe657323fac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-08 12:35:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a15901d2d152158bc860de0d3a924156a9d76822","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-08 13:55:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1877cfc34407443fc6511db7c6aa040b9af1a7ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 14:01:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a58390915b2459d8c2082b83fda5d0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7a8e1e19f8bd4212a6d87f4c6a129b45 : SUCCESS in 31m 00s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b93ea8603ca74bb993fc734e2a585eca : SUCCESS in 1h 07m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bae7af97b75042118f49e5f93cbcfccf : SUCCESS in 17m 56s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad41776988d54c6d982255d6384fa384 : SUCCESS in 56m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a723f9998042404f85924ec07509fe72 : SUCCESS in 1h 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f08e22690464ca686c4b948f85b30a9 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc51820f3a9c4e1faed729e7df75308e : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0944ddfe5f1e442eaea1763e26549be8 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a9528fc41454c989ed2a434f2c85c86 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fed88e4f36864800ba081edb6bd9c1b3 : SUCCESS in 4m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85ab1998ee904daa92adcbdf2410799b : SUCCESS in 20m 11s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4e72616ce74047218ea8c34951c64024 : SUCCESS in 1h 00m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/21f4c1d8b87e49e497ae6b2fbb4de1dd : SUCCESS in 28m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e756aa5d43ce483a876240c9c949d2bc : SUCCESS in 49m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/017f9072fbd3434fba090ef79f714bed : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8885ed29ddf557170d9754424109b096cd05e534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 14:01:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6200e16d5a648925900df1238880434dece65f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 14:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed44def0459493aab53b8fe6afde944\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7dcb6f9e3e2e4b4cae95591f4bdc7576 : SUCCESS in 26m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d54a31abea343f09c43a2817c21dd74 : SUCCESS in 34m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/15839abc0b8840998569a56d400b1893 : SUCCESS in 1h 11m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6af9a1c0b5d4a09b621a6ddca888611 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7478047fbb3457b8ed9c1daeada4b1b : SUCCESS in 4m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/63d4cce56af84628b027909cb67bd378 : FAILURE in 1h 56m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e27b65af4618432c942d69344d9d4bf3 : SUCCESS in 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac91c5d69854309a9146bfe423f9daddb92136da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:42:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4670edd4dfda4c41a5699728e326fd4e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/79a6f3ef5fd54ef28b9b80960fea12cb : SUCCESS in 19m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed3e17b78d7a41d3ac0046e45fcec85b : SUCCESS in 1h 15m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/70fc67f08794435ca11f6c21db0e14ce : SUCCESS in 1h 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d99d01603c445981a2d29a77942b5b : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/323e94246db84a0a8dd543e4623c1098 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09cade18da01446496b0c88e2b2420bb : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3c254dfc79413a8b358ff9494be76f : SUCCESS in 5m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f26fb9689ee44a629953fe18a027444f : SUCCESS in 20m 23s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/aadb073c0d3f41b6b05322aebbe9197e : SUCCESS in 1h 13m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4fd8e76917984ab3bd4e25e84aec0499 : SUCCESS in 31m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bdc8520945e4a90b5cc514f1a0a96ca : SUCCESS in 43m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ecef69c1986f4583b2ac8312618dd940 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6c29df0bb0cbd6da8dadb9b22fe5655f11903970","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:42:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2031d925ebef69acf1235c16908ddbe32a61bf26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:43:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d0c8f28a9bf4f4692ab7d08688ef5d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8ba7d36cb14b2fbb0b8393ef8cd4f8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa95f9686588c1b4624fd30c923787a555e304b3","revisions":{"fa95f9686588c1b4624fd30c923787a555e304b3":{"kind":"REWORK","_number":1,"created":"2023-02-08 12:35:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/873003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/873003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/873003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/873003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/873003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/873003/1"}}},"commit":{"parents":[{"commit":"fc8a908c0011ad0c1020149a6e76e9cbef269e43","subject":"CI: Install Apparmor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc8a908c0011ad0c1020149a6e76e9cbef269e43"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:49:53.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-08 12:35:10.000000000","tz":0},"subject":"CI: store only templated out Dockerfiles in work_dir","message":"CI: store only templated out Dockerfiles in work_dir\n\nChange-Id: I0832d7997536f5f0dc195d5038545b6cc220f0fa\n(cherry picked from commit c45a742e6fbfd3392511b315edde17ab42a9760d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa95f9686588c1b4624fd30c923787a555e304b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa95f9686588c1b4624fd30c923787a555e304b3"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
