)]}'
{"id":"openstack%2Fkolla~873894","triplet_id":"openstack%2Fkolla~master~Ie6c078a9c37f54d60f0d5a4aadb647e184489cb8","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-15 13:58:57.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-02-15 12:21:11.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-02-15 13:58:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie6c078a9c37f54d60f0d5a4aadb647e184489cb8","subject":"Pin setuptools\u003d67.2.*","status":"MERGED","created":"2023-02-15 10:25:40.000000000","updated":"2023-03-06 13:30:08.000000000","submitted":"2023-02-15 13:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873894","meta_rev_id":"0cdd395a03d56345c02ee7be680a2085617e4bf8","_number":873894,"virtual_id_number":873894,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 13:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2023-02-15 12:21:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-02-15 12:10:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-15 11:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2023-02-15 12:21:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 10:58:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-02-15 10:58:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-02-15 11:53:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-02-15 11:58:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 12:10:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8eaa027c03b12a82f241fdb3ab00eb027ccbdd34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-15 10:25:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ac92ccb56f6fdc7df6d077d3979892185c5ab1d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-15 11:53:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f917c3c156fe83e72e78c614ccb5a8c2904f7e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 11:58:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4f0f668d564690b77d5f5b20c5596d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f29c29c290fa44d6ac1729d54c87e98d : SUCCESS in 25m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ad213e370fc04c54990353d208f64736 : SUCCESS in 1h 25m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5f4e22234d4a4df69c170dec48f78cc1 : SUCCESS in 21m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c497daa42a4c4421bf6f98a340b6f143 : SUCCESS in 1h 06m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e2951e1e9f4b4da4822e601de5ad6ca8 : SUCCESS in 1h 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb62fc449664913ae76cfa807dadf50 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e769693db94b7c971eca52910bfa2f : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/812c45e973aa4e51a72dd0fda6d4b585 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/826dffafe2a54ac0846d09187ed806b2 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff4e66cf17e471bb029427c24909125 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9bc6612d1eed46cd82cceb4b91a985a2 : SUCCESS in 17m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7efc2fc4286c4860bc9faea2666a63ca : SUCCESS in 1h 09m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e1ee60a5e3ad49328ff02a02470a66b3 : SUCCESS in 1h 19m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/21fae708ce454e629b668b3517075096 : SUCCESS in 16m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4989dd2932e24f58b284e00801c9ec5a : SUCCESS in 47m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1de9db5cfd6f416e8ca8e80196d549c6 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2b5cdd96a9e3f7feb7cd11ac45ece6aa713061d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-15 12:10:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b2468c68dfc31383b90ce0dd4126e268b3213b7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-15 12:21:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41cf8e1da61fc1f18678edfce91fc6c317725e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 12:21:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8db58c749bb6e7e023d04cc34d8a4794fdc7ba3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 13:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9055725e308b4571b8d403d6a4ba49c2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ae55bcf1684b43d38a14f7b1d7eddbec : SUCCESS in 2h 56m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3621b4f41a5f4c90a0b27807a58b9404 : SUCCESS in 2h 10m 17s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afc23b36afd1401ebba02d1044a87b9e : SUCCESS in 2h 44m 33s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd22af874df54797891288ae16492116 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc8a43ab5dfd4a9c856125e859f0cf6f : SUCCESS in 4m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9f456b4fca5f44189c0be98d8ec363c1 : FAILURE in 2h 18m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/17bb67fe51f64225ac27690a1fa7a72a : SUCCESS in 1h 51m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17c72448b5f7e34b102ccaa21dbb90ef0556cb08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 13:58:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a25ad2b8183405fb5802ad01c34e0a0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf5e49b53f2b406d8e005095e3cfe0cb : SUCCESS in 18m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ceb6da3be52e4d1f950eb3956efaf699 : SUCCESS in 59m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/63d23fe64fab4c2eaf42c40bda28ba1b : SUCCESS in 1h 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00bbd5fb33ff444b8eb4999e9ad93217 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbe64d38c1024196a8764591a4b90ec8 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80b13d0be6644dddbd0ddfa5f7e6f9c1 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1628e8405454b93bae505b2cd1f4764 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5c7e6d2eecb245ecb4a4ecf8f3c7239a : SUCCESS in 18m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/46053fc81226429f8ad4d90b0e8281d8 : SUCCESS in 1h 14m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bd0ac1648f5748e6a2583e9aa6573986 : SUCCESS in 1h 12m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/54875478dea34f41adae6aa5caa9f06d : SUCCESS in 17m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eb72e0c5ac6745ca971ee7552a0fa3e6 : SUCCESS in 50m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/65bf387d690f4a2b932bb7d47a100d6b : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0deebfeec81cc4ccc9f22ecb8e589a66fa174bc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 13:58:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"abb3680953bbb50abd03772c4b8bbc8a9694e0a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 14:00:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/745aa5af4df44f3381d1d5761240f941\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11fe1bf11d0843e4a79894ba449e8031 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1},{"id":"0cdd395a03d56345c02ee7be680a2085617e4bf8","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-06 13:30:08.000000000","message":"Created a revert of this change as Ia4ce65b0c250360ffa5d7221686656763a400276","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca0238598a37c4a174c7b8d55b64c15bb34ef58a","revisions":{"ca0238598a37c4a174c7b8d55b64c15bb34ef58a":{"kind":"REWORK","_number":1,"created":"2023-02-15 10:25:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/873894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/873894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/873894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/873894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/873894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/873894/1"}}},"commit":{"parents":[{"commit":"c1cea309a42034e32a18840201383f77e7da31e2","subject":"Merge \"Bump Opensearch and Opensearch Dashboards to version 2.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1cea309a42034e32a18840201383f77e7da31e2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-15 10:21:57.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-15 10:25:38.000000000","tz":60},"subject":"Pin setuptools\u003d67.2.*","message":"Pin setuptools\u003d67.2.*\n\n67.3.1 is out and it broke Horizon builds\n\nChange-Id: Ie6c078a9c37f54d60f0d5a4aadb647e184489cb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca0238598a37c4a174c7b8d55b64c15bb34ef58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca0238598a37c4a174c7b8d55b64c15bb34ef58a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
