)]}'
{"id":"openstack%2Fkolla~873964","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I70e2bb1d2744d38e08f5a9a79ab6a7265ebf24c9","project":"openstack/kolla","branch":"stable/yoga","topic":"opensearch-kolla-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-16 16:21:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I70e2bb1d2744d38e08f5a9a79ab6a7265ebf24c9","subject":"Add Opensearch image(s)","status":"MERGED","created":"2023-02-16 11:04:15.000000000","updated":"2023-02-16 16:22:56.000000000","submitted":"2023-02-16 16:21:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873964-opensearch-kolla-stable/yoga","cherry_pick_of_change":830373,"cherry_pick_of_patch_set":25,"meta_rev_id":"687bebdaa0d62d2c45aaf8e67c3a5df89cb752bc","_number":873964,"virtual_id_number":873964,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 16:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-16 12:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-02-16 14:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-16 14:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 11:04:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2023-02-16 11:28:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-02-16 12:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 14:25:51.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"dabc9017e76d3c363ed33b4dd83dd1b336eb8c16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 11:04:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/common/users.py","accounts_in_message":[],"_revision_number":1},{"id":"7d82cba1300490d14f24f11096297a059f8565b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 11:05:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"19c27583ef2600c7c02e611809b2b7a8aec04a52","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 11:05:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d9d24683be25384d9e5b678e9af765dc9ae81aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:50:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9a0adb3f01c4ae0a85419e8d5260f94\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d9ff5625a3e44b08d0fe1e6fbafe849 : SUCCESS in 51m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e01d8e3b2d8f42558bddf9224e3bf447 : SUCCESS in 42m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3cff83cf533a4024ac88e97d3564682a : SUCCESS in 1h 19m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bb233b317fc54b428697f0a7beac6cd1 : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4861e50a1d484d4084d35225c42a8bad : SUCCESS in 1h 33m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3aa2ea7d4aff49e098f66902b49869bd : SUCCESS in 42m 32s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/838a52da8714433fbc7c5e62ed994124 : SUCCESS in 45m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/47b369048d294706abb6aaed6a9011f9 : SUCCESS in 1h 12m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b2bc03afa99a45ada08647726d2b6c9c : SUCCESS in 1h 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d2801e70a140389244fe0972c67af4 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0792c01258f428f8122a467f048e0bc : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dc87a1353694ad0b3dd7b67ec446880 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4785ebc8f38849c4bd09d0bf0b6a730f : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14cc2cd32a9c4078ac94779ec2ecce20 : SUCCESS in 3m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6580ea821ba643f4800e864c9f5215cb : SUCCESS in 29m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/980ddaab454f45779edbfa42d1625194 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":3},{"id":"dec3ad34618f5f9f5eb8a03acf01f0e9be2dea31","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 12:54:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"4acee6237462f55f9560227d2fd23435f5a1d88e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 12:54:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5f8fa4dd3b11e766485ab63b8ce441f3ca866d6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:15:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9042b621a99425a948cf664886181e4\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad375ee5230e481a978b7bbce781d9d9 : SUCCESS in 2h 55m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68b4633e1d07421a88b2d61586aebd4d : SUCCESS in 42m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/691903d7edac46398b9d07701dfb18bd : SUCCESS in 1h 23m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eeee1d5e1bf343838aaa056a808ed708 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0dae6f0669d841e4bffc714bb8278788 : SUCCESS in 4m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eef36da542574fab8437ac763dee0584 : SUCCESS in 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0470be4c87343bbaf65fc28f50c0db631a1dbe1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-16 14:25:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf6fc87d5230270af842215cfc5fba043460973a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:26:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc07825d11a94a715f366a49b0917ab1d9227e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:21:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd0dbb3237f48bb967eefba712b9157\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ad76f8b0892246b993ab978a17f88681 : SUCCESS in 34m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/492f7ca0304e4878ab31bd9a96de5135 : SUCCESS in 1h 07m 41s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ac48c8ef13fb411b8fe22bd42902d50f : SUCCESS in 1h 53m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/11115201ca2a41caa9e4bd6d368be003 : SUCCESS in 36m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f4a7b3f9c8a24965ba90c8c92cb0aa56 : SUCCESS in 1h 10m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/32dede47c1ff4d8eab33f7505a1db0c8 : SUCCESS in 1h 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceeaeefab8b24202845276b1b0d94bea : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b113506e7c4a41668ac0783a9569196e : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e30f4e5316b4e5189f60026345a17c9 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20c7d3027b04487a40bd9370838eb2e : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4da88ca542af4bc69d85555d4c67194b : SUCCESS in 4m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23ca4bff1a534f32be10b7363a0422f8 : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":3},{"id":"e49ab86ac6d326f0f023507c0fc00ed1f4684793","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:21:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"687bebdaa0d62d2c45aaf8e67c3a5df89cb752bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:22:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e80606549524578a70594653c7b8c58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dab8d02d632465bbff307009cb15a88 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2070ced1457a4a7f96d05caf0488507b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9050b67252ff199b8fcfebc2af8203053c7a452d","revisions":{"97a396be784a5babc5eb77c629fdf93612a05951":{"kind":"REWORK","_number":1,"created":"2023-02-16 11:04:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/873964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/873964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/873964/1"}}},"commit":{"parents":[{"commit":"48308c7bdf1e3059f3557dd7560efe2e2ea4971f","subject":"Merge \"Add util-linux to centos base packages\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48308c7bdf1e3059f3557dd7560efe2e2ea4971f"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2022-02-22 10:58:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 11:04:15.000000000","tz":0},"subject":"Add Opensearch image(s)","message":"Add Opensearch image(s)\n\nHandles x86-64 and aarch64 targets.\n\nChange-Id: I70e2bb1d2744d38e08f5a9a79ab6a7265ebf24c9\n(cherry picked from commit abfb5713ae24799bc7fc2d5b007348fab154c4fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97a396be784a5babc5eb77c629fdf93612a05951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97a396be784a5babc5eb77c629fdf93612a05951"}]},"branch":"refs/heads/stable/yoga"},"5adc6f3296e4b7932c6712f41ce770ed76eaa69c":{"kind":"REWORK","_number":2,"created":"2023-02-16 11:05:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/873964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/873964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/873964/2"}}},"commit":{"parents":[{"commit":"48308c7bdf1e3059f3557dd7560efe2e2ea4971f","subject":"Merge \"Add util-linux to centos base packages\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48308c7bdf1e3059f3557dd7560efe2e2ea4971f"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2022-02-22 10:58:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 11:04:59.000000000","tz":0},"subject":"Add Opensearch image(s)","message":"Add Opensearch image(s)\n\nHandles x86-64 and aarch64 targets.\n\nChange-Id: I70e2bb1d2744d38e08f5a9a79ab6a7265ebf24c9\n(cherry picked from commit abfb5713ae24799bc7fc2d5b007348fab154c4fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5adc6f3296e4b7932c6712f41ce770ed76eaa69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5adc6f3296e4b7932c6712f41ce770ed76eaa69c"}]},"branch":"refs/heads/stable/yoga"},"9050b67252ff199b8fcfebc2af8203053c7a452d":{"kind":"REWORK","_number":3,"created":"2023-02-16 11:05:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/873964/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/873964/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/873964/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/873964/3"}}},"commit":{"parents":[{"commit":"48308c7bdf1e3059f3557dd7560efe2e2ea4971f","subject":"Merge \"Add util-linux to centos base packages\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48308c7bdf1e3059f3557dd7560efe2e2ea4971f"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2022-02-22 10:58:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 11:05:23.000000000","tz":0},"subject":"Add Opensearch image(s)","message":"Add Opensearch image(s)\n\nHandles x86-64 and aarch64 targets.\n\nChange-Id: I70e2bb1d2744d38e08f5a9a79ab6a7265ebf24c9\n(cherry picked from commit abfb5713ae24799bc7fc2d5b007348fab154c4fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9050b67252ff199b8fcfebc2af8203053c7a452d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9050b67252ff199b8fcfebc2af8203053c7a452d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
