)]}'
{"id":"openstack%2Fkolla~874055","triplet_id":"openstack%2Fkolla~master~I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-21 17:26:34.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-02-17 08:44:38.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","subject":"CI: Retry docker push on publish jobs","status":"MERGED","created":"2023-02-16 07:24:19.000000000","updated":"2023-02-21 17:28:22.000000000","submitted":"2023-02-21 17:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874055","meta_rev_id":"3375e19db4f4c33f10f9a9ec73ecda9eb17f0438","_number":874055,"virtual_id_number":874055,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 17:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-02-21 13:34:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-02-17 08:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-02-21 13:34:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-02-17 08:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 08:28:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-02-16 08:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-16 08:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 13:34:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"c6df05d494abf372fcc47b916e7a8e2900bcdb23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 07:24:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a4b8f3ca87d47b94dca3cfd133b4e100caa7942","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 08:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d8af163e5b4fe38dfc1ed0059784ef\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6df5efcf56c841cb92f81939db7dcd04 : SUCCESS in 32m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f30022ece0d43b3a15638707dabfc12 : SUCCESS in 24m 04s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/579f9d9b632441a1bda970819ed79b80 : SUCCESS in 1h 16m 04s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9d297496fce4644a22adddc8688ed92 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9fa3d57d9cf465baf0d508670e483af : SUCCESS in 2m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/53f56e782c434167babae3d95a9e81f1 : FAILURE in 26m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d4f4aea17ed4daba4cc1db2cfc4c1c9 : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f597b95515d416bad42cbcc811dfa943e1e6806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 08:59:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfa9c60be1dd4b87850de01c05f030bf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/29ba5a6c28214e23b09b2cc8aae03ccf : SUCCESS in 18m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8822d6adb5c9428b8299c8dc1be5e07e : SUCCESS in 1h 22m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1e82ffb51d4542c8b96820e1b30c6a0e : SUCCESS in 26m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e53b5f6e3898490d92f5971c36880e8b : SUCCESS in 1h 11m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/209cbdcf11584072a8d856ba0007074b : SUCCESS in 1h 29m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49256a9c5d44538a9cb5c9741f78d34 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac41d4bb123b4e2d8004fd475f352549 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d764b7a1fbf4d21b0d51dc144c27da0 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1d318b3675748ef8564fac70ca9ed2c : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c01378a90b4d0abfb97f52cb48f3cb : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3383cdb964c148f49d34b2c4526f7cd0 : SUCCESS in 20m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0fba8dab5736448f855a8b0df420370a : SUCCESS in 1h 15m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5baf7d36cf784e54a78dace73d8666a0 : SUCCESS in 1h 21m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/71c4201a24e845a9a2c115273dd79b87 : SUCCESS in 16m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c147df3a103a4d5f934e4d3ce257eff0 : SUCCESS in 46m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cb2a92e56eac40f9b4c911e90a0aa260 : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cd5997233f0fd285c40f735dbc095a4b4ccece03","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-17 08:44:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3712379a2a908651dff4c5c1e91af3679e8929e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:45:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ac953fb5fbf609c0186eec773ca28f0d130828e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:24:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb3b4b1818fc41f1a5bb1d181e63b76c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c08eefea03af44eca4f6b3ab05607edb : SUCCESS in 23m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/300c1f0a41bb49c28d02cc44c35132f4 : SUCCESS in 59m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f1a96e86993046fea57628a4f6b89d2c : SUCCESS in 1h 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe20ff01ffa43bdbd10d243e922bad3 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18b49324aff74da7aff97754e3131c86 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1bd64bc318948ca84de2088a1b8376f : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4fc0e40dfe4a40a1f56c3b4f25cab5 : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/556dc103fac4485a9c433b9332b8f2f9 : SUCCESS in 18m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f9c76b9fd9cf4c9bac9b9630bed21dad : FAILURE in 41m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6353e6b5b75c4b5e88b63eb85b162dc0 : SUCCESS in 1h 14m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/66fbe2c2adf645a4b317f9d1c9899513 : FAILURE in 30m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2be8f531a84b45efb0de6e94536e0bb0 : SUCCESS in 51m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b1395790a1324619b0dfdf4fe001398a : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2ceb896ef173ec3f83522c0df76e121e72893083","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 13:34:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73c8f219d33f687a5eb8532bb38577c90a20f054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:01:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/411581855d7e4ad39b6368aa352c3d36\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/39c5b205594347d38e71aa852a8675fc : SUCCESS in 21m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/886155942cd7486bab5e5a8f34404f96 : SUCCESS in 1h 16m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/68f2e8c70f394f22b38536d00f36dcd1 : SUCCESS in 19m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b22a6284e439474d91d809a23de1ee2c : SUCCESS in 1h 08m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c522d5aa7b2a4705b2bbc80cc58a3381 : SUCCESS in 1h 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c6e066d9da4aa6ba84932d1981b41f : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5acda324c3af4b30bbad6c795a397179 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a343c215d05a48be8f780b31b7cb8ce1 : SUCCESS in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94d18bfa1e4741afa5ff5ce2bb7c69f1 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740408c344ae4ce2977a2e5e29a8231f : SUCCESS in 5m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7d383184f6c644399cd41ade6f6f31df : SUCCESS in 21m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/48dc63c48ddf42918b8684bdf5b86c63 : SUCCESS in 1h 07m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cf62fb35f63e4f34829fdd41b7913159 : SUCCESS in 1h 20m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/31d5f398be284423b8f1789ecd9e6d6a : SUCCESS in 37m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cb9de00c6ebc417c944b51b70abcaecb : SUCCESS in 1h 08m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/613bb7d1bad942ecb7bcb0442f892331 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dd7e3746ab9b1930eafab7fa647e994f46916237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bca9eb61b8c8918030d1bde736e722b19b13fe47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:26:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60405311af5e4ffebbde6c2b5f792a69\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e128bc6e0821418685e22044aa2f85b4 : SUCCESS in 15m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/571d9b2c753c4681b10999f71b2677a9 : SUCCESS in 1h 06m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6c1fbd92502749208cb3ea598ca92416 : SUCCESS in 1h 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507d2afbcf9a4cc89069754bcf350d19 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/079ecc0ab0354a6d977a940bae1637f3 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fa45598f60d4151bfb4bc401f09118f : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d1005a092441f39238efcb8a64e499 : SUCCESS in 5m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4f00d3a8cba04d28baeb416936bf1550 : SUCCESS in 22m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7ac7ffecfb8243fb9585dfbe362926c8 : SUCCESS in 1h 10m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5f6d8ae55a8a4d7c91d7d06b14e7742f : SUCCESS in 1h 18m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de2bfb60eca543049b3308735e42a737 : SUCCESS in 15m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d2552fc8256c4d2b9406c7cec115eade : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/024a4a7d6bcd406e8c9e336d9465a6a4 : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e2ff0fb4b382a4514ee07f1b90b8170b81a35d13","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:26:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3375e19db4f4c33f10f9a9ec73ecda9eb17f0438","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:28:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab6ab674a42b40d4962accbdfbf16cc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a551f08027bb499a8c028f6f6346b2b8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0887d2e789e8fcd41b49c6346034c4774ea919dd","revisions":{"0887d2e789e8fcd41b49c6346034c4774ea919dd":{"kind":"REWORK","_number":1,"created":"2023-02-16 07:24:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/874055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/874055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/874055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/874055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/874055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/874055/1"}}},"commit":{"parents":[{"commit":"ca0238598a37c4a174c7b8d55b64c15bb34ef58a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca0238598a37c4a174c7b8d55b64c15bb34ef58a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 07:23:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 07:24:16.000000000","tz":60},"subject":"CI: Retry docker push on publish jobs","message":"CI: Retry docker push on publish jobs\n\nWe have a high number of POST_FAILURE jobs failing on at least\none container push - let\u0027s fix that.\n\nChange-Id: I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0887d2e789e8fcd41b49c6346034c4774ea919dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0887d2e789e8fcd41b49c6346034c4774ea919dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
