)]}'
{"id":"openstack%2Fkolla~874282","triplet_id":"openstack%2Fkolla~master~I524132d55e2090787aa6dcadd989b7b1fadaf986","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I524132d55e2090787aa6dcadd989b7b1fadaf986","subject":"Do not install weak deps for rpm distros","status":"ABANDONED","created":"2023-02-17 17:27:51.000000000","updated":"2023-02-18 13:15:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"e2f48389eea6d9b83749aa340038444206e79665","_number":874282,"virtual_id_number":874282,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-17 19:05:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-02-17 19:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3ac2449e4ef2220ec169283c2d86ecb386e6399","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-17 17:27:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36897adf437698ba2f32f9b0e6d8548ea38266f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 19:05:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4690ed964e7c46e881785a8e2e97cdc0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0ee8a5021f5e4a04ac11c631a75f4128 : SUCCESS in 20m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c4a72ae3347e423ca53a6cc4d65d4ed3 : SUCCESS in 1h 15m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6766dcceeea6454c9e46321ed2dd2f08 : SUCCESS in 19m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ca8135e72ee84c498446a77ff96eb019 : SUCCESS in 1h 06m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/deb4147278424625813556ebc76f3352 : SUCCESS in 1h 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa54fd50da8483a991ca02aa2183fbf : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3b47b20085c47a28a7d00653b292e82 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/781ad6aefe7841eba1c1a1b7c7357fda : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df2ed8a18e80496f9a546508f5f9ac3f : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11768dd446a749a4a71b3a5c29b6ef6a : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e852c96280754d408df9eab96e60e6ce : SUCCESS in 20m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6c89bb23b10f49248d3c0d10fd44ddce : SUCCESS in 1h 06m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/313df3b6d64f4fe68c89f66b18c3a63f : SUCCESS in 1h 21m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6864a21eeb1046d6ae6eff4f2a1386e0 : FAILURE in 32m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35cc7bb0c15f4eebbec2e2510cfb8506 : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4c231232c1c445bd9b972ea53266aef8 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"03f2d625b746b075b4c554412ab435573ad7be59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:39:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/813be64c8f524dd281dc4912391f91cb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/11ff7616237143df86af98919243b388 : TIMED_OUT in 3h 05m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4c873b8df83b49368ff4ba38a1ac9dbb : TIMED_OUT in 3h 06m 06s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fadeca87aa094c3285ff28ca79541664 : TIMED_OUT in 3h 02m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2dd9accffdfb4a9c9ecd59ee8048fa20 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18a090cdcf644633b8a924b01d18a2d3 : SUCCESS in 5m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d89f5f60e9e346d8b4b22a62cf7fcc2b : TIMED_OUT in 3h 07m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d1c17161f68e4095b661a5430d5f3498 : TIMED_OUT in 3h 04m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2f48389eea6d9b83749aa340038444206e79665","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-18 13:15:12.000000000","message":"Abandoned\n\nactually it\u0027s already there","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbb32b61f642b3870383aee74cd96132ccb0b11b","revisions":{"dbb32b61f642b3870383aee74cd96132ccb0b11b":{"kind":"REWORK","_number":1,"created":"2023-02-17 17:27:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/82/874282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/874282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/874282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/874282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/874282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/874282/1"}}},"commit":{"parents":[{"commit":"ca0238598a37c4a174c7b8d55b64c15bb34ef58a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca0238598a37c4a174c7b8d55b64c15bb34ef58a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-17 17:27:28.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-17 17:27:48.000000000","tz":60},"subject":"Do not install weak deps for rpm distros","message":"Do not install weak deps for rpm distros\n\nChange-Id: I524132d55e2090787aa6dcadd989b7b1fadaf986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbb32b61f642b3870383aee74cd96132ccb0b11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbb32b61f642b3870383aee74cd96132ccb0b11b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
