)]}'
{"id":"openstack%2Fkolla~874383","triplet_id":"openstack%2Fkolla~master~I0d072ae6081a9427aebf4dc562501a0a6ef160d0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-20 13:48:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I0d072ae6081a9427aebf4dc562501a0a6ef160d0","subject":"bifrost: mark unbuidable for Ubuntu","status":"MERGED","created":"2023-02-20 10:35:27.000000000","updated":"2023-11-22 16:08:49.000000000","submitted":"2023-02-20 16:34:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874383","meta_rev_id":"bb5484a58e8098323fe000f3e364db881105c86b","_number":874383,"virtual_id_number":874383,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 16:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-02-20 12:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-02-20 12:01:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 11:51:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-20 12:01:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-02-20 12:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2279deb77b953aae0c311fe430a5a84c76ca533a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-20 10:35:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0d371643c14e2342cd306a35e92e6b801ed8d8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 11:51:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5329513d6f6d4a838193854b5753a9b8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/acc3f9eb26af4a2fb3d1d644b14a9cd2 : SUCCESS in 30m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d6bd5ecf7ba0482295378ae8f6838edd : SUCCESS in 24m 04s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/23e04594185a43e9a8b02248e2dda83c : SUCCESS in 1h 13m 49s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22f0d0f1296844a4b3e18901dc0667b0 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/019d83c286a540879ac61dced6748aa7 : SUCCESS in 2m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3b0daebcf5cd42d892ad71fbd4f266ab : FAILURE in 24m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d40b7e3cda744737a38b37740622b3bc : SUCCESS in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc158c8c40259dba680e1f70364866684521aea8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-20 12:01:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c0658a43e30e3f60b9c351f3c4b07c5604cd815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:14:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb5bef07e6e459986ae1dab618465e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/39741c06a15d4e6793fa3fa41a4cc1ce : SUCCESS in 20m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/502b51dd444a49908009b48f561ffbe5 : SUCCESS in 1h 12m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3d68e39c1cbd445ca9125f97c40529d5 : SUCCESS in 15m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a9f5a14723ca4257977ba19c8d4f5fd1 : SUCCESS in 1h 12m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b8699d32864748cd8ec1d4630155470e : SUCCESS in 1h 22m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a3a3c7bf7042d8b824337361a72c29 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b9c8e7599384e03885d62ec310fcfec : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3fa519cd45541a39d62c7df2d19341d : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bbe134938f645059b67df8020881d7e : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f1128a495b46fea6c2775f7cb876a1 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a08f482026344899830d3597a44eba28 : SUCCESS in 20m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bbf3fb4fe83643cb8f975402acfa71f7 : SUCCESS in 1h 33m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7d97d6891c974d7984a09f2062866abf : SUCCESS in 1h 26m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1ba4b854a20046a48e0b4a6d9382922d : SUCCESS in 13m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/81715ceb75c243dcb338434082825b00 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f9de6ddd1b894717949d424ba87ff657 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dff63db12b0dd6e29215e0fe00a8050c71b951f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:14:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5490bdfb6735a2303e8f98987486775838c6a06c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 13:46:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8aa591338fd42e5b2774a6849cfef64\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5f9739c8175945eea0d656c50535ddad : SUCCESS in 17m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff2158cbdf9f49f182d68802b4beae43 : SUCCESS in 1h 04m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bc793f8164754513994026aecc164ee7 : SUCCESS in 1h 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c5fdc3c96844dd95d121830d94d821 : SUCCESS in 2m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4673bbd975b34417b8e6602e8ba3dd7a : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8da26289a1e46c388a06168e79b2900 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7130364b8945d487b6e2cfab72e155 : SUCCESS in 6m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/20007ea8b69c4bbebea26c03cfe4f9ca : SUCCESS in 21m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/19ce2bcd36504c698d9af21c9ec19d07 : FAILURE in 49m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/69bbd8a88fab4119949cc751fe8297cc : SUCCESS in 1h 13m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/52365d60f3d64744b0ccd63053d4c19c : SUCCESS in 10m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/717593209b014f0f97d455ec30fc5795 : SUCCESS in 36m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ba98c89b5e8e4b559ad0fec8319d1e87 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cf9bade2c34801e8afcc2d9632e5fee9d9760874","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-20 13:48:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73b0afbac542518e1acf7de39017b6c2d411f0e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:11:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53cf1ea310474f64a91bcc3c3b21a38d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0f20556ec469455589fdc633b6161bde : SUCCESS in 37m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9a468e1dee1647439a6a64af3a03b831 : SUCCESS in 1h 10m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/258a16ced1f44164922f2ce861f18ac4 : SUCCESS in 18m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3196e8e656c8417cb25520e762f37c98 : SUCCESS in 56m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ce38f4fef8d9459c89152faf6adbbb3f : SUCCESS in 1h 03m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac0f2004c83a4b898763d81a62128dae : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/722346af7b144484bfe559c3300a4bd0 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7608de91606f420ea5133363516b791d : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84a7353d05f64cf59afe7fbc81a9476b : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a958d00e8a9349449ecd2c20aff37058 : SUCCESS in 4m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/45666863eafa4bd5ae61a8011194b65b : SUCCESS in 23m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d920c31ef49e422e9d1afb98370c2379 : SUCCESS in 1h 06m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d4400449434d4cfe92d7db8fb4bab9f4 : SUCCESS in 1h 15m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72d2d8ff3bab4c3e9269c65f16b728d1 : SUCCESS in 13m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4379e0b445714cdf8d20f1d2903c26c2 : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d84ba72f471140629285fee1c064dca9 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":1},{"id":"51361e6922f6651c34bde51d6ee2e32aeadd79f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:11:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fe04ae2358c6790adf621cffff96634437ca33b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:34:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ee82239aec4198a517de605acbb2c6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/766a84440148487fbc21c92a02dc02bf : SUCCESS in 17m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9f938862db884007ab5dd58407bdb359 : SUCCESS in 1h 00m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e55504cc44574ac99d2e544f346f6936 : SUCCESS in 1h 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c181bb4a92e84069b86e49a03eadfbe5 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4caf0afc71cd4bcb87aca5c840d1bf19 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9658c5f629bd45568f374f71203e5651 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14fbae19488146fc9519ef9fbc9efd5a : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d9d3e9892f544e0a9e9fc84280ca72cc : SUCCESS in 18m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b0fae83f001645708f4da7c5e4e3689b : SUCCESS in 1h 07m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/07d50557446a434dbb11ee7a2400a4d6 : SUCCESS in 1h 16m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c019b5fd63fe486ca3d1987c33edd5f9 : SUCCESS in 12m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a52166debcde46279186b1c074622d04 : SUCCESS in 1h 02m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/21a833dacf3843918fba68f368c80561 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f64c331d7085b4c05ae3c7dc2c1fcd0e8af27d32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:34:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a38cee0e37a810a1ac993aafdbee1ee2d6d39ea2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:36:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dcf836bbccb42a0a65d499163c85862\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2171d2476a74f46b6cdb2d83dddfd4d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"438f76914c201af5d5fe5a340a18d87b8fe1bbc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 16:55:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c42d16b2e1456c8411324a09b62e58\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/48d379e36cbf4fedbe0692f6b5d5761a : SUCCESS in 2h 31m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cd95f37f90554070b35354129d4b4155 : SUCCESS in 27m 58s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5bbf5ef294384886a853517120843e00 : FAILURE in 55m 04s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a20f005d1c38436f9fcf9923d3b0962d : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/272e53fbf1d544ad97d42c84c5428bd5 : SUCCESS in 3m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f1154ea4c7f5446ba5aaa85418ec104f : FAILURE in 3h 01m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/49a2951f2944402a9b2e84c95d771b1b : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfd4ef31ae6d66b0bf6e529d441ea1f690e17de3","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 07:53:58.000000000","message":"Created a revert of this change as Id409c2b5423791d4ca9626e376de07ef7c1a8d8e","accounts_in_message":[],"_revision_number":1},{"id":"cea111e3400b83aede50a278d1d5b020bfcc97dc","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-29 12:48:27.000000000","message":"Created a revert of this change as Id77785c34bf046a679e9a7afc1a4591ed2a4138e","accounts_in_message":[],"_revision_number":1},{"id":"bb5484a58e8098323fe000f3e364db881105c86b","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-22 16:08:49.000000000","message":"Created a revert of this change as IId77785c34bf046a679e9a7afc1a4591ed2a4138e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2f747fbba7adf3059d0da9bd899788b4a45763c","revisions":{"b2f747fbba7adf3059d0da9bd899788b4a45763c":{"kind":"REWORK","_number":1,"created":"2023-02-20 10:35:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/874383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/874383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/874383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/874383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/874383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/874383/1"}}},"commit":{"parents":[{"commit":"ca0238598a37c4a174c7b8d55b64c15bb34ef58a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca0238598a37c4a174c7b8d55b64c15bb34ef58a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 10:34:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 10:35:25.000000000","tz":60},"subject":"bifrost: mark unbuidable for Ubuntu","message":"bifrost: mark unbuidable for Ubuntu\n\nUbuntu packages have changed efi shim file name and bifrost is\nfailing.\n\nChange-Id: I0d072ae6081a9427aebf4dc562501a0a6ef160d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f747fbba7adf3059d0da9bd899788b4a45763c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f747fbba7adf3059d0da9bd899788b4a45763c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
