)]}'
{"id":"openstack%2Fkolla~874408","triplet_id":"openstack%2Fkolla~stable%2Fzed~I179c41032239f5f0ce6055c79295b32c33fe67a5","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-21 20:02:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I179c41032239f5f0ce6055c79295b32c33fe67a5","subject":"CI: Configure additional disk when available","status":"MERGED","created":"2023-02-20 18:16:30.000000000","updated":"2023-02-21 20:03:38.000000000","submitted":"2023-02-21 20:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874408","cherry_pick_of_change":874375,"cherry_pick_of_patch_set":20,"meta_rev_id":"b0681207799e2a5559c7ce80e13c12d4dde757d7","_number":874408,"virtual_id_number":874408,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 20:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-02-21 16:17:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-02-21 13:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-02-21 13:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 19:46:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 16:16:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-02-21 16:17:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dfb1934d6b61dc61c6045f05127218f8a1d05549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-20 18:16:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e28ccf1f3d1108c56851f054ac5a5d4b39cfeb74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-20 18:32:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8960c3b9b1190831cee0cd571e389a6ae15c844f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-20 18:34:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d76c4131c952c573de3362f865f19c77484f0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 19:46:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a802062f96d4a96a266632f1218bad9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/907518c6576a425c8e860162a06cb495 : SUCCESS in 26m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e74d13eea34341f58bf63c41de5e6f8c : SUCCESS in 1h 09m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/426b59dc624744d7b5c170e00de2f49b : SUCCESS in 26m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ea5d3df2609c4ec69daddc21edbca882 : SUCCESS in 55m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2d73fe5889ed47af9ac04a06d6afc7c2 : SUCCESS in 1h 07m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0160e236f84058824ed4b4d329722f : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd26c29ed7c1419bb43da933d2c1cb15 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25ef09686cac4c4aaadf8270509cbd90 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98086e281cbe46bba8d4acd2280021cf : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78ba86fcaec4f0ba38d05b116dd3959 : SUCCESS in 6m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ab9289f40fa24ddebcaba2d9c24d510d : SUCCESS in 20m 45s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/dd96102d688d4bc1a4dde595d5d20ea5 : SUCCESS in 1h 05m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dd6a132bb09d44d4a1f92346819dfef1 : FAILURE in 39m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d418fd52931448c188a1b98370c3dfd6 : SUCCESS in 45m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c0ad169a6aa044b1883ddd22e27223f2 : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9b4f0fb5773d4d570a844822670fa2829b1035cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 21:01:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93e0bc577fd64d0b8732de95d2cf0e7b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f660463f5039434081f6641bc0022b0a : SUCCESS in 29m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5e4dfe709cca4380baa3448722730dfa : SUCCESS in 33m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04e3cb1d1fd241ac95300d64da26cb33 : SUCCESS in 2h 25m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/348b27336c694ddbb908cbfa24f20635 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dbd7fb7060d4760a0c5ff9ef448c691 : SUCCESS in 8m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/49b3a28a08de43a6834e46522855bd07 : FAILURE in 1h 53m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1534e800ade745aaa1f9404f9c54830a : SUCCESS in 1h 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fff8be69fa2c628e66804e8eb82cee2a531f989a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 10:35:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3c418c2149807d6e0bb792dc106188b8536ed741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 11:28:24.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"83ddd716bf9897524bbca290a48bb92a7ff02cb1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 12:33:07.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"d777788d7f91f9689642d5517ff31f983262908b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:36:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fde5ea43f34faeb9b770c6229cc20b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/889dd33b9f5e43aebdae06acb8f6e3f4 : SUCCESS in 17m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/cb086c5dd2eb4686a7377b555fec2094 : SUCCESS in 51m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c7facc9a834842c8835c1befc8ac14a7 : SUCCESS in 23m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/410b475f77fb40eda084cf0ed7c9bed7 : SUCCESS in 51m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/43dac2f2b72842c89623f731c4370a86 : SUCCESS in 1h 07m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c927366245417099529976a2da1a99 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01bf71c589e34f56941002a426c6124d : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9800162b34ff4290985cf588f210f8c9 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe0b88c0cd5041b8a35b9afef4c34d69 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb065946afe2449390e7394bc5b1de56 : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9d652d7facf147efa43ede65a883bcb9 : SUCCESS in 24m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/890c1e42e6bd4156958a5f5fb8b3277e : SUCCESS in 55m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c52d79ee70624c7b8ddad85897e54c64 : SUCCESS in 15m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11f19152705644e2af85d44ff2ebf6ca : SUCCESS in 40m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6b7bd6f970640be9c447bd776dca9ea : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":4},{"id":"09cfc2a124e86b4d6bb02aa3d10a22f50aba4441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 12:41:53.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ba55777bb6fbd23999be122646cb76d423ce3221","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:41:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6efb9a675acb469d83f2c64313cea24c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1f7eb49c7b964989aaa70894a8beaf65 : SUCCESS in 24m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4b9888defdc64f528b610acd52014123 : SUCCESS in 24m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e38ca7f48e848daad53c80e799c6a9f : SUCCESS in 1h 11m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25fb2f064d8b400d905301ea915d23f7 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/943148422dc64fadb1e236b1e8359fb7 : SUCCESS in 3m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4d3174c4e5054a2884cf80023028c5d6 : FAILURE in 24m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3d416758612c472ebacac9d21e6fbe69 : SUCCESS in 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ffcc48348318b368aa659f996f0e206082281992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 12:42:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"58d112eb07d3bb6de54aaaf783e0ed8b3e4fb95a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 13:02:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"37a1a05d3af0a49049ecf3626edc2de8deea07b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:08:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b1580837401431ba87c660144948e8d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cd0499528b834c4e89eb9756dc98393c : SUCCESS in 26m 44s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b0eae5da6ed94a758fc32ed2e0d0b768 : SUCCESS in 1h 18m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1b48f26dde5a4d62858d78b9afc02eac : SUCCESS in 21m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0b82c83d651b4801a9a107243220eef1 : SUCCESS in 59m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/faf1bee4ac1e42299eb32177c77b5cf9 : SUCCESS in 1h 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd32a2b8460d4dccbcd5141cd6976ce5 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be88fc2542a4f8fbc29b4f8d80cb057 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/056cd1a8266b4af2a28f435c58fdce28 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bae523349bbc4b44af291a1bb09ce057 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f12a506bd0469f8cc76b8313310b4c : SUCCESS in 8m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1a5cd608d59c49618e49a931b6439f33 : SUCCESS in 28m 10s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a7461fe99a664baeb1819f672f98220a : SUCCESS in 1h 04m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/577238bba1af47c68e4f65f323ef011b : SUCCESS in 13m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc1bb64eab76415ab1362d0eb3dcb3d5 : SUCCESS in 50m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8f90de793c80402f91ca9e27957ba661 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":6},{"id":"d773ba43347f746ab695a9be71b1598f4a286b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:08:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fea81572c8bbdc26751f2cf9d1686baab76543ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:00:16.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/874501 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"a224e7444e0e21bff5580ceb35658979bc6e4840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:10:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21b48851d35e4c44851bc259b8d487fc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3b40b88324f845db984fc8108db5b621 : SUCCESS in 20m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e5abc804dd140c8b183f240dc44bc3d : SUCCESS in 21m 43s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0021da48e8fc44cda19d8c0fcb33fe50 : TIMED_OUT in 3h 02m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9bc60da06f24dd3a324f7e2f3ad2548 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3870258772b14980b7cd39593811b9d1 : SUCCESS in 6m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bf6c26e845ea443fa306b5fbb47e417c : FAILURE in 18m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/955a5665d6564427882091734228ca37 : SUCCESS in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c42716a98ff4b5187ba4f7cfac9925af50505714","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-21 16:16:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"494c6b2939dc4b85072168d2672ea337fe827fae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-21 16:17:13.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"edfcf6313f15e89ae07eb882739f79949bad0a46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:38:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ba7e1f19724626bcf92f9810741a5c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/103b616cf3e642b490192f299d330488 : SUCCESS in 30m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e57085fa651740d1b70e1637c1b6cd6c : SUCCESS in 32m 16s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32b292dfadae4b079a27b6ce8e85583a : SUCCESS in 1h 18m 24s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/136bda39ba904756b5205c6bc39f2f61 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c60f34efcbe24927a7623f95bd3d39c6 : SUCCESS in 3m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/44a8a8fcab1f4084a9b292eb15fadef8 : FAILURE in 31m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1ee8afc9bf284689923c38b57936a0f3 : SUCCESS in 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d090001a6c2c8d686aaa658e7ca07b31cbe3207e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 18:30:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc43e4eecad547bfb37432b6b1d6882d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2ce703ddb2c6492f8484e441f65da7d1 : SUCCESS in 25m 02s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/14088f0a60924df190c26daf5c76e8f6 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/962b26acd7574c019ee149b338075e0b : SUCCESS in 18m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cb2e41e01db143c5be69260d500249f8 : SUCCESS in 57m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e2c6666a768842629cf45a7442e74f80 : SUCCESS in 1h 04m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34aeb0b3e6c749d6b04557010cad65f4 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4abb2c7551aa46f0b9c6b63e6ebe37fe : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a25208d85214473bbe660b64b79041c8 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63c893033c6a47d385253b88a75325b3 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6d88f6658942d7b34ed6a0bd44c197 : SUCCESS in 5m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e6cdf4e3fcd140228f2d828e70f0dfd0 : SUCCESS in 27m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/077120ec78844e06978ae11268cdb743 : SUCCESS in 1h 58m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6d8b8bbea06f4972b17682ce6abce2a2 : SUCCESS in 14m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e12f6d4625514995a1bcbf535c56c4a0 : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fb7d92d6d264d8fbf4afba688850e7e : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":6},{"id":"941aaf70fc213e42888dd081a531b3114852efcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 18:31:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3b74ea5324633f39b3d899ca16609357deb13688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:02:40.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b102da323af47759dd704a23af92a4a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a2705d9e1ad545f1abe9216247e51965 : SUCCESS in 24m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38810a5724594cd890e9871b2d099c44 : SUCCESS in 1h 07m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/62b03e3e9d404410bba4589d9aebd80b : SUCCESS in 1h 22m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fa47e8359943b8bea9f80f3f58bcdd : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a707a2757484035aa83034bcb45894d : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e27dc8b3cb15433c8cd24d7c890cf67f : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8b7959f67e4a718d55cf17ce0d05f0 : SUCCESS in 5m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/94c9417046c148a9974dfb65cc7569bd : SUCCESS in 25m 20s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/645f495fb46e4cbe8d6f81a5e7cb8d05 : SUCCESS in 1h 07m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e50089c8682b43bea4653db30114ed06 : SUCCESS in 18m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/40190d9aa1884463bca050be6ac2f765 : SUCCESS in 45m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2412cfbbb669428ab4929cefe1f303f4 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":6},{"id":"44426e5244f3f2e74cce12de57fcd0429097a8aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:02:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b0681207799e2a5559c7ce80e13c12d4dde757d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:03:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4839a0a7f37f4874be6b04c196fe8b0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ad6b0432934362bbf5b49b6bb754a1 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"278167179c39da821a1580be397eaffc0ba8471b","revisions":{"85b48789812db30f24a582e9130ae162ce51365a":{"kind":"REWORK","_number":1,"created":"2023-02-20 18:16:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/874408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/874408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/874408/1"}}},"commit":{"parents":[{"commit":"23ea06f2e8d5dcbc6237920c41d38fedbc8b4c1a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23ea06f2e8d5dcbc6237920c41d38fedbc8b4c1a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 07:43:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 18:16:30.000000000","tz":0},"subject":"CI: Configure additional disk when available","message":"CI: Configure additional disk when available\n\nIt seems that on some nodepool providers (rax for sure),\nwe get 40G boot disk and 80G additional disk, while\non others we get 80G boot disk only.\n\nChange-Id: I179c41032239f5f0ce6055c79295b32c33fe67a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85b48789812db30f24a582e9130ae162ce51365a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85b48789812db30f24a582e9130ae162ce51365a"}]},"branch":"refs/heads/stable/zed"},"9b6a19575a989928112f6dfc17ebb166f9468b9f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-20 18:32:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/874408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/874408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/874408/2"}}},"commit":{"parents":[{"commit":"23ea06f2e8d5dcbc6237920c41d38fedbc8b4c1a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23ea06f2e8d5dcbc6237920c41d38fedbc8b4c1a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 07:43:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 18:32:54.000000000","tz":0},"subject":"CI: Configure additional disk when available","message":"CI: Configure additional disk when available\n\nIt seems that on some nodepool providers (rax for sure),\nwe get 40G boot disk and 80G additional disk, while\non others we get 80G boot disk only.\n\nChange-Id: I179c41032239f5f0ce6055c79295b32c33fe67a5\n(cherry picked from commit a0d758019a753596567128a44032165dc8214f05)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b6a19575a989928112f6dfc17ebb166f9468b9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b6a19575a989928112f6dfc17ebb166f9468b9f"}]},"branch":"refs/heads/stable/zed"},"28d7b512fff2fccb30a2518b62e8bba4d80f61c2":{"kind":"REWORK","_number":3,"created":"2023-02-21 10:35:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/874408/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/874408/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/874408/3"}}},"commit":{"parents":[{"commit":"23ea06f2e8d5dcbc6237920c41d38fedbc8b4c1a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23ea06f2e8d5dcbc6237920c41d38fedbc8b4c1a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 07:43:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 10:35:47.000000000","tz":0},"subject":"CI: Configure additional disk when available","message":"CI: Configure additional disk when available\n\nIt seems that on some nodepool providers (rax for sure),\nwe get 40G boot disk and 80G additional disk, while\non others we get 80G boot disk only.\n\nIncluded fix from I06e262cd48d33ccfc0634589c82dcac80dff51af\n\nChange-Id: I179c41032239f5f0ce6055c79295b32c33fe67a5\n(cherry picked from commit a0d758019a753596567128a44032165dc8214f05)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28d7b512fff2fccb30a2518b62e8bba4d80f61c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28d7b512fff2fccb30a2518b62e8bba4d80f61c2"}]},"branch":"refs/heads/stable/zed"},"34162d1aff0e90419fe72551a03d86db51837fd8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-21 11:28:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/874408/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/874408/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/874408/4"}}},"commit":{"parents":[{"commit":"47431082f65d8761a74df40ed18ff3415722ad69","subject":"bifrost: mark unbuidable for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47431082f65d8761a74df40ed18ff3415722ad69"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 07:43:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 11:28:24.000000000","tz":0},"subject":"CI: Configure additional disk when available","message":"CI: Configure additional disk when available\n\nIt seems that on some nodepool providers (rax for sure),\nwe get 40G boot disk and 80G additional disk, while\non others we get 80G boot disk only.\n\nIncluded fix from I06e262cd48d33ccfc0634589c82dcac80dff51af\n\nChange-Id: I179c41032239f5f0ce6055c79295b32c33fe67a5\n(cherry picked from commit a0d758019a753596567128a44032165dc8214f05)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34162d1aff0e90419fe72551a03d86db51837fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34162d1aff0e90419fe72551a03d86db51837fd8"}]},"branch":"refs/heads/stable/zed","description":"Rebase"},"86af5c25c67f89f308b63340b844552cdfbdc3df":{"kind":"REWORK","_number":5,"created":"2023-02-21 12:41:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/874408/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/874408/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/874408/5"}}},"commit":{"parents":[{"commit":"47431082f65d8761a74df40ed18ff3415722ad69","subject":"bifrost: mark unbuidable for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47431082f65d8761a74df40ed18ff3415722ad69"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 07:43:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 12:41:53.000000000","tz":0},"subject":"CI: Configure additional disk when available","message":"CI: Configure additional disk when available\n\nIt seems that on some nodepool providers (rax for sure),\nwe get 40G boot disk and 80G additional disk, while\non others we get 80G boot disk only.\n\nIncluded fix from I06e262cd48d33ccfc0634589c82dcac80dff51af\n\nChange-Id: I179c41032239f5f0ce6055c79295b32c33fe67a5\n(cherry picked from commit a0d758019a753596567128a44032165dc8214f05)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86af5c25c67f89f308b63340b844552cdfbdc3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/86af5c25c67f89f308b63340b844552cdfbdc3df"}]},"branch":"refs/heads/stable/zed"},"278167179c39da821a1580be397eaffc0ba8471b":{"kind":"REWORK","_number":6,"created":"2023-02-21 12:42:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/08/874408/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/874408/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/874408/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/874408/6"}}},"commit":{"parents":[{"commit":"47431082f65d8761a74df40ed18ff3415722ad69","subject":"bifrost: mark unbuidable for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47431082f65d8761a74df40ed18ff3415722ad69"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-20 07:43:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 12:42:32.000000000","tz":0},"subject":"CI: Configure additional disk when available","message":"CI: Configure additional disk when available\n\nIt seems that on some nodepool providers (rax for sure),\nwe get 40G boot disk and 80G additional disk, while\non others we get 80G boot disk only.\n\nIncluded fix from I06e262cd48d33ccfc0634589c82dcac80dff51af\n\nChange-Id: I179c41032239f5f0ce6055c79295b32c33fe67a5\n(cherry picked from commit a0d758019a753596567128a44032165dc8214f05)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/278167179c39da821a1580be397eaffc0ba8471b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/278167179c39da821a1580be397eaffc0ba8471b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
