)]}'
{"id":"openstack%2Fkolla~874588","triplet_id":"openstack%2Fkolla~master~I6e66aea5dc0af61fbc074b4c8ed3910435c98dbd","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-28 17:46:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-02-28 19:10:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e66aea5dc0af61fbc074b4c8ed3910435c98dbd","subject":"toolbox: Increase galaxy timeout and retry download","status":"MERGED","created":"2023-02-21 14:14:11.000000000","updated":"2023-02-28 19:11:31.000000000","submitted":"2023-02-28 19:10:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874588","meta_rev_id":"0867d8594d111fe714e741ce89f507a5f48f4c1b","_number":874588,"virtual_id_number":874588,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 19:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-28 14:38:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-02-28 14:38:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-02-28 14:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-28 17:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 15:40:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 13:29:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-02-27 02:27:42.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-02-27 09:34:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"655e34dc64ae0659951ef6a5a9bc89abb431583b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-21 14:14:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b194fb40e8ca8c8274ea6b111a6b582ad50a1fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:40:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a8740459604ab8961e3d0b5863eb5b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a81d98b62c5e4b0785dea87ff7402298 : SUCCESS in 23m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6782994d0abc40df87f3bccad39714ce : SUCCESS in 1h 07m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d70cb8dc7a0041f7a0127db8a8712245 : SUCCESS in 22m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ea5cb5e0d23648cfb03f523af0dc9caa : SUCCESS in 1h 07m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/61d504c008824d9885049ffea47ac72d : SUCCESS in 1h 08m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8867f20b0e7b4f9fb7847fb58f339bfc : SUCCESS in 2m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87c53eb2c9f640609a3798836da3eba0 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/964f87c72b0442feb0ff5f0cb9fd02fd : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/521db886ff2448be8e9745e0e449092e : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f43ffb9341f487f9a606f27cf9aac99 : SUCCESS in 7m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/05a977b1ad8748baa3654f1a87d806db : SUCCESS in 20m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a41b7e6f6e55435798dbe339419c0f75 : SUCCESS in 1h 13m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4dd8c158684443bf9287350426297fb7 : SUCCESS in 1h 12m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3c767dc094ee4f1e834c01603f1cb49e : SUCCESS in 14m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/af40009c51df495e9449f37b5d4aa66a : SUCCESS in 46m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bd5cd597ba034761ac23d4614e404767 : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a29d62adb2d9cff62217169f5dadd7ebb9a5e5c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:20:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721e6b52722d4da480098831855b4b02\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c659a5cc88d7483c90d5a14101a208c8 : SUCCESS in 3h 01m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a60ae434f21d48a38757cff4a520e0e1 : SUCCESS in 26m 59s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e3e892a39ca407d946675d38618c636 : SUCCESS in 1h 21m 16s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71884d37c3d3404ba01baa5c4e3f4cce : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f306d31277dd45f186bf5417763976a4 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/846493fb3e454ee2b5d138bddd1e4ad7 : FAILURE in 29m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2bbf5617b7c94adf82fabdbb6c4202d3 : SUCCESS in 2h 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"459c67052c0a0d71dcf1267b47fb3c8d3ea18dee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-22 13:29:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91cf589e66525dddd51b0908d1a082fa97601f67","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-02-27 02:27:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9639cf2a54b947b54ee35db30d982309ea3540be","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-27 09:34:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d72c7f9aa01479d3ad47fc0c6880c2a8be8a5061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 09:34:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ed08e96ff7b835256e04a54d931c6d9ad947c18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:05:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba564225ca684ffba9f2183f3a1158a5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b81eacb1496648e2b0fabf25a0d7a361 : SUCCESS in 24m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3eda44aeb9b24efb9c17198cf71b290d : SUCCESS in 1h 03m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b0b8a165b4b423fb26e627a89949640 : SUCCESS in 1h 09m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19fbce6a371e46158819048c54599854 : RETRY_LIMIT in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/990c5df15f8844d1bd6100ccb413bc7a : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8895d1ebf3f1414cb1bd530de52ab8a7 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d7b3203dd24938b942fa672bf0856f : SUCCESS in 4m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e35202b23220433c98efb983fd9fa382 : SUCCESS in 29m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4ccfd6f3248b4015bab94ea521366cf0 : SUCCESS in 1h 13m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b85a47f39ad54a079d47a322ed726441 : SUCCESS in 1h 24m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/496b0a18d2c3498fb0eac02c94317530 : SUCCESS in 22m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ebb7d4ef16d040c1b7af3a4b586d2d59 : SUCCESS in 49m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/124d8cc7699541d59951e28475c86490 : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b2c5a65cc5bfeb0917531692a939ed4e94d90c2a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-27 11:08:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"562a7e9d2d17423c912df97d88e6f6e1c3c7739e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e200a2539d54525bf04e867015da9b1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f310d8b109df4d3d857d9570840810c2 : SUCCESS in 19m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bc45865146914aa0984f209ab83155ef : SUCCESS in 1h 19m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b524a0ab8efd42758955c6c26c9c13bf : SUCCESS in 17m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/35abb5a20f8c4d6ebcaee086ddfbf5f8 : SUCCESS in 1h 03m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8cabf93c9ddc495d92f4009627898767 : SUCCESS in 1h 08m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ec2c0729a8457292d431e7c5798b15 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1ae2197f1a64384888d27c5e878f618 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a53c0002e024c27a941c04acb0ae081 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91c5d3c63e2b4ed9b9955183b888c4c7 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c0f4270b514fce909d6b1ea14fa432 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/65c2811355d84992bd13651451bb3779 : SUCCESS in 19m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a402b27b134a40fa9f2c8a97c5233fbb : SUCCESS in 1h 10m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9bc5eae53e5e4123ae9498a926bc483a : SUCCESS in 1h 21m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/96719de8a6a6433cb130e439b39d43e3 : SUCCESS in 13m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/325220f753ef46468a551e9c09dc48ac : SUCCESS in 41m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/82faef058461456294e28e9ae22b0dfe : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c2ecf477a138a84cf997f746d1adc2b9b10693c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:48:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ac543b0eb7edb856a7342d2e8b47119b811414a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:40:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e0a576eb3847dca87d3e254b5d348a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7ae2b80fe8104c029b300fbf8a6dcfd5 : SUCCESS in 1h 42m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ad9e6961cc9f44ad81a7f3f265f3d71f : SUCCESS in 1h 30m 58s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/690f1ff603184caa8c98b830ae93b321 : SUCCESS in 2h 28m 57s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a279ae9691e64165a5cb0393d5878c8d : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0feb00d19df24a64ab35cb8af35b94a9 : SUCCESS in 6m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/db45464dbcc541999d4426eab20a4be5 : FAILURE in 1h 48m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/36afa054c3044add98b34455de76043c : SUCCESS in 1h 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8b2c903fdce81db8b477e94a90d170435bbe499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 14:15:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e7a4fa88d148ae818d52177d378ee3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca7ce49c6fdf4e74b7d7d57216b6a082 : SUCCESS in 13m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/76bd6648c47b4bec81d280add2599ff8 : SUCCESS in 1h 00m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/74a3cfc06d5b4b979cff5f228e3a6845 : SUCCESS in 1h 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eab783e467e4b25932d473945952f3c : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59058b933d854bcdbe5fe9740177c9a5 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47a4423323034e9ca042400a8a3e89ef : RETRY_LIMIT in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d19abf5e6ef4e2bbd0b291c9089e875 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/46bd17022d464972aa86e0bdf4739b10 : SUCCESS in 21m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2a1c8ba5290448869be1c08a2a17ed4f : SUCCESS in 1h 07m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/deb7e93867ea4c4283b8447fa4f6dab9 : SUCCESS in 1h 22m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6fa403b88b884c3e907bc083edcd684d : SUCCESS in 13m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/90a39e78132c4886a6b08b773b631b74 : SUCCESS in 39m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/95bb28d7583143e49b86179b11ec842e : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"38e09deae8ed8aeb4038d54e3ca45cc0c86f139a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 14:38:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4db891db321b9e0d114d51e11e9f3b9f8cb1eba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 16:12:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b37ec5d4a434f96bab59d58fd93789a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/12cc94c938e24e2196a1ed0c22bd222c : SUCCESS in 24m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c09327b9a8c14fffacdead62a8203445 : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/862cdb7638f84e848962b942d7f833f3 : SUCCESS in 20m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1e0a0aa209594ac6b5ede6ff0bb2c31c : SUCCESS in 1h 00m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/105386f2139e4254a124192a59d0e36a : SUCCESS in 1h 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f501a50a509e4597ba77521f7e41c11c : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/596af2eb8a19439e831c6cfe60758bac : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f83468d33214eda93af5b5f321f30e8 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c5463e79e0241fc94d8a8c01714120c : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e024a6a6d83e49afb31415fdcf87d3dd : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/959db19784624c708ccda52b07eaafc8 : SUCCESS in 24m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/988f3eab183f485dbbe42361d5d121c8 : SUCCESS in 1h 09m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bb644657a21d4f67b560d874490ecb8f : SUCCESS in 1h 19m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4f5845ff08b74415b40a3988397d3a2e : SUCCESS in 14m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/40e9788302cd46d4b3cd6caeace4e16a : SUCCESS in 51m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2159b03682c049b0935b71b3dd183c5c : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1deff1cf5e1d1772e1315c3368efdfc6d21192fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 16:42:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53bd0bda78043078cd56c5e6c27cb8f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cc22edede154479eaabb05d53112835d : SUCCESS in 17m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fcf47daa864a4ccfb72e2035318fc730 : SUCCESS in 15m 07s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/99e84185d9824ce1a8024753ff62385a : SUCCESS in 1h 20m 38s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/519d9943868b4cf99c2616c7c36a1347 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52f200a8536b46209442a50ef2ea6658 : SUCCESS in 2m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f6dafb11bf1f4476b037aeb97a910d96 : FAILURE in 2h 01m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d5e94811b91f47deac80ab3033de6af3 : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62fd53d37ad9f9d7bab014758b01fd3ae779ede7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-28 17:46:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6e2107954a9e27cc4317550255b993f6359a8bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:46:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8238e0516891ed8ce704e8febfdf59c4a9439074","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 19:10:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ae032f6b054c4e92cfeda9a4fc1741\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c9b91d3a67e4496ca47faafd368228a1 : SUCCESS in 19m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f597c43a7463438086c2c8c54f31c75b : SUCCESS in 58m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a5b775a0b3884e72b3d83f46cba3129d : SUCCESS in 1h 02m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04013b5055284a85add392fcaf49c504 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8719bf55e6b646d6a97bd33dc7472804 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0706bad8b9e04472a6c8aaa5267be772 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461d59f3c9334809a2a7606f73c98fbf : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c008cb36787e446dba602ac6bc088a0a : SUCCESS in 22m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ed7285d762b34567a45e240f8f1d2fe3 : SUCCESS in 1h 06m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/878dfc7dbd1146ea97b12cae8678974a : SUCCESS in 1h 10m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/380859bbd04945dd8c4353d2d49d3970 : SUCCESS in 13m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cf4bdc89b8e44ac3afffcab68388e16c : SUCCESS in 53m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/142c013e02d346d3874236a4ebd016c1 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c2d5c9dabfd1253b7f3771f6fab64e2925829a5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 19:10:27.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"0867d8594d111fe714e741ce89f507a5f48f4c1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 19:11:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1601b85f49544861bdfb0db6b6b31261\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a619f82975d94ff4a041ac15dcce93ab : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"575a090af222a6e07520570690e63a9ca7af5434","revisions":{"82816cc8b277e1417cabb7e41916236ba01075e2":{"kind":"REWORK","_number":1,"created":"2023-02-21 14:14:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/874588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/874588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/874588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/874588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/874588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/874588/1"}}},"commit":{"parents":[{"commit":"a0d758019a753596567128a44032165dc8214f05","subject":"CI: Configure additional disk when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0d758019a753596567128a44032165dc8214f05"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 14:13:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 14:14:08.000000000","tz":60},"subject":"toolbox: Increase galaxy timeout and retry download","message":"toolbox: Increase galaxy timeout and retry download\n\nChange-Id: I6e66aea5dc0af61fbc074b4c8ed3910435c98dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82816cc8b277e1417cabb7e41916236ba01075e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82816cc8b277e1417cabb7e41916236ba01075e2"}]},"branch":"refs/heads/master"},"575a090af222a6e07520570690e63a9ca7af5434":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-28 14:38:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/874588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/874588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/874588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/874588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/874588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/874588/2"}}},"commit":{"parents":[{"commit":"35088cf281da2c883b1444fa6cc76013ceba5eac","subject":"Merge \"Add a way to add other container engines\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35088cf281da2c883b1444fa6cc76013ceba5eac"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-21 14:13:55.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-28 14:38:10.000000000","tz":0},"subject":"toolbox: Increase galaxy timeout and retry download","message":"toolbox: Increase galaxy timeout and retry download\n\nChange-Id: I6e66aea5dc0af61fbc074b4c8ed3910435c98dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/575a090af222a6e07520570690e63a9ca7af5434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/575a090af222a6e07520570690e63a9ca7af5434"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
