)]}'
{"id":"openstack%2Fkolla~874627","triplet_id":"openstack%2Fkolla~stable%2Fzed~I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","project":"openstack/kolla","branch":"stable/zed","hashtags":[],"change_id":"I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","subject":"CI: Retry docker push on publish jobs","status":"MERGED","created":"2023-02-22 07:37:49.000000000","updated":"2023-02-22 12:56:31.000000000","submitted":"2023-02-22 12:55:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874627","cherry_pick_of_change":874055,"cherry_pick_of_patch_set":1,"meta_rev_id":"9e790f6da2e5d983f5e20654847b03eec5a7dde0","_number":874627,"virtual_id_number":874627,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 12:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-22 09:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-22 09:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 09:15:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"461da366bcac741eb44756b333a4f9483408bb79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 07:37:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6ae11372f0cea77ec4dabbe5fa1f7a75c4f318f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 09:08:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1977bd91c1e601b3f998de184fe726e98d6f370d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:15:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0b14bc8a9441558392e020fe9d8c68\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c72942b46331428eaa04587d58f28320 : SUCCESS in 22m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3bf76f82d18949b58b3d1a292b49b87f : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0db474ca1e864df39e14e2d8e11eb873 : SUCCESS in 22m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc13cf58f39d4a9c9ff0f28c1ffe2c8f : SUCCESS in 55m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8c13567d97044b78a45bd554a097bfab : SUCCESS in 1h 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58955ec1885f4a0db72c0ff5e3dc7df2 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e9f0a78f21c4495a6d1a0f77e0f04da : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fe845f72fa04882890aa7d36f375d69 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a587ee11a86c449d9027a58ed4bc4093 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52268f147064a5a859a242edbbe681f : SUCCESS in 5m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ffcea88278a34cc1b64b82ea91512190 : SUCCESS in 19m 17s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/620d6525f67c448899e991ed4409fbfe : SUCCESS in 1h 07m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2cc5c04752454c8cb8168f8dce55eaaa : SUCCESS in 32m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e10d882cc97b4a34a5eca3d01ef60f40 : SUCCESS in 1h 01m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/712952c91bbd4dec8aa95f1359b19ebf : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":1},{"id":"36cb1505dc1310e3181e0c1c44c3297b1eda9015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:16:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26dfbd2e385d1bfc36dbad15b854b0d54265af79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:18:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46c76c16780b47e2b8a4817d74752507\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9c24ea064f4a4a0a9438a245bf33a268 : SUCCESS in 37m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/474adc79e1f34b75b732067c44d2ce19 : SUCCESS in 36m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/776dc52fc6fa4501b7b5024a91fa2bc5 : SUCCESS in 1h 25m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b618f923f180497f92db6731e38c205d : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec7544b659484691a890716cc9a21341 : SUCCESS in 3m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1a5a8eb4c6c643f3a91aa38099d7a87d : FAILURE in 1h 37m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f0d564aad8924db7ac23e9c7de2c6e76 : SUCCESS in 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdd854ef8dd1a7bb30e6bbb0f6babaaa0aa3131a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 12:55:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33808004e61140cd911428520773b76c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9cf951cd328d4cd787573fc629e0d8d7 : SUCCESS in 23m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/05afc34c6c534dbc8cae8aa177d9fee4 : SUCCESS in 1h 01m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2f4113821dc844bf9ae059900158c87f : SUCCESS in 1h 08m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f858f454b447b38674ba6c663e7fd6 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b2ae2f90ebb4036b12e278fa441591c : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6db2bff436041c0b51826ab801dfab4 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde18eb07f4a4be2a9e957dc271f8553 : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e50a2effda7a4be7a34119d6fc43a818 : SUCCESS in 22m 19s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/97aac0ae74cf400f85626aae5beaf314 : SUCCESS in 1h 10m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6441723bf6304824bfe389221bedf2ad : SUCCESS in 12m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/178d6520a86743aaa64a4ff228d7eeb0 : SUCCESS in 38m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7ead0dc4db449148c0ac574963630ec : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cfec285fd134b5c37fb9a3c65718f2c2d7a5d892","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 12:55:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e790f6da2e5d983f5e20654847b03eec5a7dde0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 12:56:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ec954e997764b29b1c0928812bfc003\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ad0fd328734e8e9c2e63577eb5531e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5","revisions":{"22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5":{"kind":"REWORK","_number":1,"created":"2023-02-22 07:37:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/27/874627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/874627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/874627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/874627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/874627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/874627/1"}}},"commit":{"parents":[{"commit":"278167179c39da821a1580be397eaffc0ba8471b","subject":"CI: Configure additional disk when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/278167179c39da821a1580be397eaffc0ba8471b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 07:23:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:37:49.000000000","tz":0},"subject":"CI: Retry docker push on publish jobs","message":"CI: Retry docker push on publish jobs\n\nWe have a high number of POST_FAILURE jobs failing on at least\none container push - let\u0027s fix that.\n\nChange-Id: I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41\n(cherry picked from commit 0887d2e789e8fcd41b49c6346034c4774ea919dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
