)]}'
{"id":"openstack%2Fkolla~874628","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-22 12:26:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","subject":"CI: Retry docker push on publish jobs","status":"MERGED","created":"2023-02-22 07:37:57.000000000","updated":"2023-02-22 16:15:41.000000000","submitted":"2023-02-22 16:14:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874628","cherry_pick_of_change":874055,"cherry_pick_of_patch_set":1,"meta_rev_id":"38fa6a7efe9de637dd0fbe7dfd2373069ce09786","_number":874628,"virtual_id_number":874628,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 16:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-22 09:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-22 09:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 09:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0f32ea970e3766ed3d183a25492f64556a70ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 07:37:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f234b7a3e5055403b7d33ff932f31df6ffd3b978","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 09:07:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb218441be5165a2e9d51cd714c4d5a8c04369e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:51:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03159dbe42e84be6b6e810b1044da1b1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/da794805c1e24794b4779714f81bbb83 : SUCCESS in 52m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/430c40870baf49798eef6a9be1281932 : SUCCESS in 58m 20s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/b75b1debc4f14a929a604b9cdf14d1e3 : SUCCESS in 59m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/14f5dcfde6234ba1a3d2af05fc04ac6f : SUCCESS in 1h 18m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/27fcacf23a4940c6aac345e8397b6a72 : SUCCESS in 1h 17m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e35df136877947ba86daa822edd043a9 : TIMED_OUT in 2h 02m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/963d43a7523a46e6ab3fc85654b5110f : SUCCESS in 1h 28m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a56c77b78ff344f18e279b14c0412004 : SUCCESS in 24m 43s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/988335cc14024fbfbf365dd6a48d254d : SUCCESS in 17m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf8f39c2cbb1444f8edd69153e42bc5f : SUCCESS in 59m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1dd061db69ba434f871515af124d0268 : FAILURE in 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a24001c4d34669a5478ef9eb4ecc2e : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28125302217e474798787df108aed5b6 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8af4ce559b8e4b389c08c32cd3cb5234 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba78fb2df35a49c79b3504de62367b1f : SUCCESS in 4m 53s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/22d95b9941974d1f91788e9cd3ed6c36 : SUCCESS in 26m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/960be0d9da844c349cea54cc7bafab8d : SUCCESS in 1h 14m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/66367c42a43e4e1ca42790090897b367 : SUCCESS in 51m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/398191a8c68f4737a9a4ff926a37ce67 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"29d59d4664f6dcbe11a0c184143ab512af5be509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025fc862051f43459468b0ae5d77adfe\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4995edd0174e4569bbd377fb42404f4f : SUCCESS in 2h 39m 05s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/77f936182c3440b8b97d85e59bc1f0a2 : SUCCESS in 2h 44m 03s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bce6558791d545b3adf17cb42daac135 : SUCCESS in 44m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8138475e6064306af861df63f7d31f8 : SUCCESS in 2h 41m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67bf57896e314e40bb1c5edf3ce7ff39 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70c6d7d0c07f44b9915f29d887206f6d : SUCCESS in 4m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7e858043e05482db32552219c0d201e : SUCCESS in 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66916bad3ec8dfab9a12ad882148084610d7d8bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 12:26:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ab7cea623fbe78d9fac966f2943b38d4b82ecef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:43:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e461846afe86488a9f163bcbfad2f1ff\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc14717f096c4cf8bf884a1c5de53748 : SUCCESS in 29m 54s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2589b3596fc4aeb8afc7f2ffb1721e2 : SUCCESS in 29m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de83ed06530a4120ac098043e2faaa2a : SUCCESS in 21m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e628114d286a4aa7ae0d0582f64fc2c1 : SUCCESS in 1h 15m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b0e56fc60a34cbbafe9d63da088eb93 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/911f5098ba6045deb25172e573a2fbca : SUCCESS in 3m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/72d28821200e41348fb617ddf1eba9b0 : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39ee5863a0e505c975a892c46aa861fb0aa02a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:07:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1b8d4489454596a0df1dc7825b8ed7\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/28f7a3ff6d8e475c924cf00adc5fa2bd : SUCCESS in 34m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6bbb32069f864e7884f8235f85101123 : SUCCESS in 34m 46s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/a7d58755c207463ea5d784914aad6ba8 : SUCCESS in 35m 38s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/320e7747698b4f83aee04ae6b3b2a652 : SUCCESS in 1h 27m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/27304bb1843141b1b1864cdf49c9ffb3 : SUCCESS in 1h 09m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d8f4304852e84f5985c721512ad84704 : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f350909fea65456fa0feb265537b42eb : SUCCESS in 1h 32m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/472a11299c9b4ea98d56dbc46af19ace : SUCCESS in 23m 44s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6433308f1afd48b0acba1033f0279d62 : SUCCESS in 18m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/006540be77c244de92f7738d93c49379 : SUCCESS in 59m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6036fcb9d10441b18e56d7e7b44f47d2 : SUCCESS in 1h 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a6a903be2c41408fa8a8b19519f3ca : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61ca60bdd9e742d0bca0bb864a55efb4 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93d14435e4d04832a8000aeb66ab27d9 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16de8ac26e234328b90a62c6995de6e8 : SUCCESS in 5m 35s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/e395017e60ce4dce96bbee724f676c82 : SUCCESS in 26m 01s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/dce6cf5d5bc34ea0acd11cce0886b1e5 : SUCCESS in 54m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf0ba8dc5d53411488c2a28e8778c7ef : SUCCESS in 16m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/87c46836c5b94314ae2d24830462d786 : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":1},{"id":"22616a20f4d57580bea6b37e53f0043dbd632dc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:07:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05c7dfcb20be63530ef8bfe78cbf44fa29b42e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:14:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d330233990dd45fe819ffc48f82fe63c\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bb44be8056cf447c8fe593c4dffca5dd : SUCCESS in 32m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/40e250487b73474892e3765425481ec4 : SUCCESS in 1h 12m 16s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/842e83f3857b4e65a2d7e0d229a78025 : SUCCESS in 1h 25m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d2723425bd8a4a02acdc2841ceb9bf03 : SUCCESS in 23m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a9158865f764a338766bc684542730a : SUCCESS in 57m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c4d0c24264cd44d99a328875b92a7e18 : SUCCESS in 1h 06m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ed1cf4847f49968dc3ea7b32151bae : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e59bf6f160941dc9151feefa4ada799 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15d2888914084f628359c2e243fe0efa : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c36a16801414fca93302de6ef700d8e : SUCCESS in 6m 09s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a4e80651a0e2478b9f9a74270ec80b63 : SUCCESS in 1h 11m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd03200b4389493280ffbb585ea6b32b : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c8006e64dc659997016b454f2a954f2ed57b32a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:14:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"38fa6a7efe9de637dd0fbe7dfd2373069ce09786","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:15:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44fbe842c1ed49e49ef2643ef361e2a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df15c1c34bd941cfa8ad981c2adeb1af : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d40f67f041240fe2acd2ecf7c1e40e14f964aca4","revisions":{"d40f67f041240fe2acd2ecf7c1e40e14f964aca4":{"kind":"REWORK","_number":1,"created":"2023-02-22 07:37:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/874628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/874628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/874628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/874628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/874628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/874628/1"}}},"commit":{"parents":[{"commit":"7a9ff37d0cd6bc1776cc5e834f0d34cb5012f44b","subject":"Add RockyLinux 9 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a9ff37d0cd6bc1776cc5e834f0d34cb5012f44b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 07:23:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:37:57.000000000","tz":0},"subject":"CI: Retry docker push on publish jobs","message":"CI: Retry docker push on publish jobs\n\nWe have a high number of POST_FAILURE jobs failing on at least\none container push - let\u0027s fix that.\n\nChange-Id: I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41\n(cherry picked from commit 0887d2e789e8fcd41b49c6346034c4774ea919dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d40f67f041240fe2acd2ecf7c1e40e14f964aca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d40f67f041240fe2acd2ecf7c1e40e14f964aca4"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
