)]}'
{"id":"openstack%2Fkolla~874684","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I5b523231cf7e1be8ce41181dda08f05e30b873c3","project":"openstack/kolla","branch":"stable/yoga","topic":"yoga/setuptools","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-01-26 10:33:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5b523231cf7e1be8ce41181dda08f05e30b873c3","subject":"[DNM] Test yoga build horizon","status":"ABANDONED","created":"2023-02-21 18:02:36.000000000","updated":"2024-01-26 10:33:37.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f13707e562ea5faf1ef4775981a1e7da5129a27c","_number":874684,"virtual_id_number":874684,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-29 16:27:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 20:22:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 22:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 13:40:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"b18230b0b1e9ceef4a3808a1ed2c9fd20f6ae6e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-02-21 18:02:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"375fbe2441458d43d1e451e6c5f9de4327fa51a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-02-21 19:12:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca7e5443ae7c6c09129d35bf4f9277d09b86d749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:22:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e59b8700e1468baf4e13fc9c33059b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/926412e6ec9742308e9113682455a09e : SUCCESS in 29m 35s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/03f297bf35e54f2d88fa50bd179d0e07 : SUCCESS in 25m 00s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c51d1d37b62478f9fc7668dff178192 : SUCCESS in 18m 34s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6eea54252f1f4a8ba3aec55084055167 : SUCCESS in 1h 08m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83854f2d95394dc19b814f506320fcc6 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05fba630b1cb46818d91eab75331236f : SUCCESS in 3m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d55f2d84fa248b89b76f508fa78f77d : SUCCESS in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f5c2e6c6dac9bc787d9c6066826a9cbe898fa3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 22:23:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3172e52d4a744aee88b60d480367fec5\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f111e4a29e744313b9e309199d48806d : FAILURE in 40m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e225b917eb2440ebb11ef8299afd317e : TIMED_OUT in 3h 02m 47s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/7ac05185fc6e4d16a112af4d6890e488 : FAILURE in 38m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/12b37b3d4dd84eaba12561901910d176 : SUCCESS in 1h 21m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9db3cf2efd0c40c0a069f6d0d0a0d83d : SUCCESS in 1h 18m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/8ce83c91b5d447dfa1bae27a2dfacac2 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d22d9adc2f29475aaf48b3285137272d : SUCCESS in 1h 33m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/db6b096309fa4e799e4e316e0fee6d6a : SUCCESS in 24m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fb408f570c9e4ffea861f671eeadca8d : SUCCESS in 20m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b8c452232c2c4f6ba9778a4a3464db75 : SUCCESS in 1h 11m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6b7bf86c7a8341acae93a0202056037f : SUCCESS in 1h 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f0c5c4d7844324b5b16aaad187ee52 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80e116d5534241e89ff80584ab8a69ef : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43459ea5e67e44bf8724b43e2c7e29ca : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4503480948874b0189fa422e683111a7 : SUCCESS in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55b17f1d5ef74c7b879e766f80279fab : SUCCESS in 20m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a8818f9aeea44789e4e4534c8822aee : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":2},{"id":"65da4e5a0853cae79490207962c7927af278d40e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-02-22 06:33:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9350226e7cab1ab750123745fcd335f87391fb3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:04:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb9beb5a2ff4a2cbee0403d3f8e60b8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff0762941a9f4e1fb270748fcc1b6baf : SUCCESS in 28m 40s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fabc8f95ecef4b05b3d9d1a1282c145f : SUCCESS in 25m 51s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed07b75da51a4016ac3c2c809c9f414a : SUCCESS in 19m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9de2ed0e359a40319f3359ecdb7f644e : SUCCESS in 1h 25m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e89310071904a979f04e5c965c529d5 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4da8c9d32384553adb5c8c65c03a2e7 : SUCCESS in 3m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc8a4378293c4f0b8dc1eaf1fa80f56e : SUCCESS in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7c454b0d017306c563b821b6939beeafc1c49da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:13:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30c4746ba8834994be63c9d418202462\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0d8c33086fd42e98a85532bf684559c : SUCCESS in 31m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ef3dc756bc041b3ae8034846547f351 : SUCCESS in 30m 34s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/86e0154bcbd14c60b629d8983a830e64 : SUCCESS in 32m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4fa516698d19489f93a165e9af5af5b3 : SUCCESS in 1h 15m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c0d78f5018cf41f994592c0da76e6bd7 : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e8dadcd08ec84ebcb3cab40c8c2c0694 : SUCCESS in 1h 24m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/09db9a2333934e10b01b6e323bc10cf7 : SUCCESS in 1h 20m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c4c97f18c70a44649f43609d99ce8ae0 : SUCCESS in 22m 53s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/20a87006bf784bf781b8e23fab05853a : SUCCESS in 18m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/067d514af78942708424c391109c34a0 : SUCCESS in 1h 02m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/090f0e7da93840d8901a075a44dab9fd : SUCCESS in 1h 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e029e28e62bb41ce8bc7ad81bf02ad02 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc9478fabc374a5bbd86bdf71586be06 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12783729bf834be2b06d8b6c2a8d0de6 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e716d1a55b4175b14426c1e5ffc191 : SUCCESS in 5m 41s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/587b467b29d742ba80baec1713a3c2c6 : SUCCESS in 21m 26s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/08846be94fd74fe5ac50294d5f6598a4 : SUCCESS in 51m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6f663afa59d4076b1b561d1fcc97542 : SUCCESS in 17m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3a3f9dfda7a4a2e99d239886624dfe5 : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":3},{"id":"4ba58a176d880628e957540f31d3eff8b45713b4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 13:40:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bcc260c0af4de071a759851dc3e61f3e95892efb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-29 13:41:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d24eba375fd93573857798a86d7dc1a37a17eb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:25:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dbf054a69654e0195b6316145b22e5d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d9129c616c4456d8d37cd4298491397 : NODE_FAILURE Node request 300-0021576823 failed in 0s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/21481f51ce184a5db45362506589e82e : FAILURE in 34m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a82380e19ed14218bb0a2a7bd60659ca : SUCCESS in 39m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72f49e5423b64ce68cadf4f5cd07c7ba : NODE_FAILURE Node request 300-0021576826 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a75d975380243179b321b68dec03e9a : NODE_FAILURE Node request 300-0021576827 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4af1eaa0efdc4aeeb6d521581c9f3e5e : NODE_FAILURE Node request 300-0021576828 failed in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d93f11265654ac58eeac30dcaca0a4b : NODE_FAILURE Node request 300-0021576829 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef59760fb73c09f168733f5cb5d3d35b99d353c2","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 15:59:17.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e011045abe047a1b72cda43f748ebb1\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/41ad13eba3f3456c843c3fbc319187bc : SUCCESS in 25m 21s\n- kolla-build-no-infra-wheels-centos9s-source https://zuul.opendev.org/t/openstack/build/ac71d18604544739a7a50492ff461ac7 : SUCCESS in 29m 20s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/5834d9d2ded44151b371db3ec76c9584 : SUCCESS in 38m 26s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/634c361e4d9c478bbdc8665cbd851b07 : SUCCESS in 2h 03m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2f50348de54a4fc2b1f95c516731f239 : FAILURE in 1h 37m 11s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/98dd15cc0a4e4c50a5fb5c8a75573cfe : FAILURE in 1h 09m 29s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/502073f730834323986cf4942d88b92f : SUCCESS in 1h 51m 12s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/0a0dd0c88a46486ba91670c3e76c0325 : FAILURE in 1h 06m 52s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/24ffa92770d543ba8f6f6d43da425bee : SUCCESS in 30m 27s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/508d8b9a59e548669a6132824a5fba42 : SUCCESS in 1h 07m 47s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/5962a07403364ddba3a960d4ac34b6b7 : SUCCESS in 1h 29m 45s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/da641e1877024e75bc7b2f65c5d5d972 : SUCCESS in 1h 32m 53s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/9751b1b6cb2b44b0a1d34f60872f5530 : FAILURE in 43m 32s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/2265b6468b5a4b02a9bf2599558a1163 : FAILURE in 35m 24s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/69548ddd642b48b1ac8292dfed33f97d : FAILURE in 58m 26s (non-voting)\n- kolla-build-no-infra-wheels-rocky9 https://zuul.opendev.org/t/openstack/build/1d61b90792c9485ca38efd2e894ad4a6 : SUCCESS in 28m 28s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b116bb4296a4419843b01fce5f9f274 : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":3},{"id":"9133f6f3d41c32f328eed3eff8e18bf20034938e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:27:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/577a79819ced4010b3702502a55a7947\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f7dcd85d262b42dea93934bbfe40b5c9 : SUCCESS in 31m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d736dedc71cd4f8ab38b019886d78b25 : SUCCESS in 49m 07s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/a3059eb55e814c2bb5fb6abd0182f97e : SUCCESS in 22m 39s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7e9d9cf6e0724f3b9358e57fff98e672 : SUCCESS in 1h 17m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/62c8dc2303f14aa986e292c4db8a5ffd : SUCCESS in 1h 14m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ce327daa3c244a7eb1ac08241ac93169 : FAILURE in 1h 09m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/801b29118c4241ecbecf1f9c19d93c02 : SUCCESS in 1h 28m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/35d57b6a863e48c4a7aa6d36acc7f720 : FAILURE in 19m 23s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4b89feb81e414fc197981ec8f964d575 : SUCCESS in 23m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d068d4472850490982b7880d6a95a7b7 : TIMED_OUT in 2h 01m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5052d78682a94712987742c16c328b40 : SUCCESS in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a3b16d7037b4346ba5ac9e20f6c2af5 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c20ad66c576c43ed96bb04bd1e41e4b5 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a62beb2e741f49b29b3d0700c4033f98 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1902306e78564417b432b4aa508443c7 : SUCCESS in 7m 29s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/19f0154639014cd78d897dbe0e673684 : SUCCESS in 27m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/5bdd2a01acd24441b79bd608a87c1527 : SUCCESS in 1h 13m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/762f11cf6b3f478e9912904677f4439d : SUCCESS in 36m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/864299915e6548db8dc2fb392c094ceb : SUCCESS in 23m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f13707e562ea5faf1ef4775981a1e7da5129a27c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 10:33:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1345ef2e670be8d7f5c2d6a60affa7970d9d7089","revisions":{"c79c131bba33efa67e526103809f980e188add06":{"kind":"REWORK","_number":1,"created":"2023-02-21 18:02:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/84/874684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/874684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/874684/1"}}},"commit":{"parents":[{"commit":"82ee1e6319d8631b2637c69f3413a020f6a27c0a","subject":"Add support for CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ee1e6319d8631b2637c69f3413a020f6a27c0a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 18:02:08.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 18:02:34.000000000","tz":60},"subject":"[DNM] Test yoga build horizon","message":"[DNM] Test yoga build horizon\n\nChange-Id: I5b523231cf7e1be8ce41181dda08f05e30b873c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c79c131bba33efa67e526103809f980e188add06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c79c131bba33efa67e526103809f980e188add06"}]},"branch":"refs/heads/stable/yoga"},"0630cc7cbea4e5ba8b966d00cdceee7e477a28e0":{"kind":"REWORK","_number":2,"created":"2023-02-21 19:12:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/84/874684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/874684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/874684/2"}}},"commit":{"parents":[{"commit":"82ee1e6319d8631b2637c69f3413a020f6a27c0a","subject":"Add support for CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ee1e6319d8631b2637c69f3413a020f6a27c0a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 18:02:08.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 19:12:10.000000000","tz":60},"subject":"[DNM] Test yoga build horizon","message":"[DNM] Test yoga build horizon\n\nChange-Id: I5b523231cf7e1be8ce41181dda08f05e30b873c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0630cc7cbea4e5ba8b966d00cdceee7e477a28e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0630cc7cbea4e5ba8b966d00cdceee7e477a28e0"}]},"branch":"refs/heads/stable/yoga"},"1345ef2e670be8d7f5c2d6a60affa7970d9d7089":{"kind":"REWORK","_number":3,"created":"2023-02-22 06:33:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/84/874684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/874684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/874684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/874684/3"}}},"commit":{"parents":[{"commit":"82ee1e6319d8631b2637c69f3413a020f6a27c0a","subject":"Add support for CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ee1e6319d8631b2637c69f3413a020f6a27c0a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-21 18:02:08.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-02-22 06:33:17.000000000","tz":60},"subject":"[DNM] Test yoga build horizon","message":"[DNM] Test yoga build horizon\n\nChange-Id: I5b523231cf7e1be8ce41181dda08f05e30b873c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1345ef2e670be8d7f5c2d6a60affa7970d9d7089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1345ef2e670be8d7f5c2d6a60affa7970d9d7089"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
