)]}'
{"id":"openstack%2Fkolla~874725","triplet_id":"openstack%2Fkolla~master~I6564886f68a997b82b49b6f975283a6d95831ce6","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-23 21:20:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6564886f68a997b82b49b6f975283a6d95831ce6","subject":"toolbox: use ansible-collections-openstack \u003e2","status":"MERGED","created":"2023-02-22 07:32:46.000000000","updated":"2023-02-24 07:21:10.000000000","submitted":"2023-02-23 21:20:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"874725","meta_rev_id":"48358bbac08baf1af8de346335f216aed84c030c","_number":874725,"virtual_id_number":874725,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-23 21:20:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-02-23 20:07:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-02-23 20:07:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2023-02-23 20:07:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-02-24 07:21:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 08:33:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-23 16:13:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-02-23 16:31:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"}],"messages":[{"id":"0c26750d94ad667df893c8ba3187d6037e723459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 07:32:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07856e3ca57851dbcaffa0dc2ee8c23e9c854e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:33:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55caac39c42844aab81235e0ac0b3eab\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6480bd494ede4552b16b4c48a65bc36f : SUCCESS in 23m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dfcc8eecbaa64597918e85a84b0c0e20 : FAILURE in 37m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7b8e08e329b45f8aa78c9e2a9dd3f3c : SUCCESS in 18m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d7e0289906714a6fa8a74104ab5150dd : FAILURE in 38m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ff5444ce942f4f1198d857a0fa821709 : FAILURE in 51m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcbf25b4130f48808f5fdd625ff96555 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d09fe98fa04b4c26bc2eb6cf2be51592 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3708205e1f143369ac5854f87276ce6 : SUCCESS in 3m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14eb65c766c44e8d92918555f0f29897 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e58ce68a5f045e692fb78d1b2238582 : SUCCESS in 7m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3217f3550bfb43e2aae692400d802a5b : SUCCESS in 17m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/569806f0155144b1b7b89282c3dff1f6 : FAILURE in 34m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/53a17ccf40f24e08a83b01e308c50fda : FAILURE in 55m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0b69e2a1bc174f698dd8154a64085c0c : SUCCESS in 18m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/310c5f8be9d946858698e38f7aab5da3 : FAILURE in 22m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c8807ed516e646f7ab37ea5894a13fa5 : FAILURE in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7cd70a585bdf5912771c3e40ca1bb60753a06dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 09:33:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3a2eeb9b2c2564ca02e88b4a35952b7569a8edb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:46:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bef3fc1e517441a0b4fa9df5929bd297\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fc8135ba669b4a31b386406dc8074528 : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/edae1d7f4a1741daa44a87f0bd8ed41d : FAILURE in 30m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86b95326f0004991963fba9e2d7a23c2 : SUCCESS in 19m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b2c4d88fa2444adca3810a876b86a3d6 : FAILURE in 29m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d55e6cbd4c2a46f786ff65be64f1c232 : FAILURE in 59m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e7b40ecd496407aa41fdd502f26df7e : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6068054561a49e9b92f2cd1e13ac8bf : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5758c9305b954d1ab615a1526a24499a : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a93c09a675824bd2bbd6d6ed1a947738 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a87a1012ab4e0a98b246316dd9eca0 : SUCCESS in 6m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7f5fb4cd77f84f5fb9cd7d2a65500ac9 : SUCCESS in 22m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d220561866b9447dbfab37058522eaeb : FAILURE in 32m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b6d337a243304ea6834ea5bb51e0abbe : FAILURE in 1h 05m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d6c6ded207a04a39ab04c6a1d44cc0f2 : SUCCESS in 12m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bdc41083de064313b040bb7f3df1022e : FAILURE in 26m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9f3a0c086a9844a7821d56440f6044be : FAILURE in 54m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f2b70b1b84852c0a81ae04b6a9fb914303aa07ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 11:10:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc60e58962d5494aae8122a66802bd44\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4af643bea3d54b9a951ac9c929690eaf : SUCCESS in 1h 19m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bac3bebbbb8543afaed5f2397dac9ecd : SUCCESS in 1h 33m 45s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/13c78c5ee5fa4707a64d7dd49a42f13a : FAILURE in 35m 37s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66e48c02b5514dc9a9347112637ad4b8 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2335a6979a8447c686865f18b264f1a7 : SUCCESS in 2m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aca98337ac0043288c7223a531ab9f5a : FAILURE in 18m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b10799cf84af43fdbc79a433564228b1 : SUCCESS in 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b10b84e8143a8ff61d37f532a0becc307a56eb0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 14:01:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2beeccf57a83bc66f8bff66555f6cf6c27763f80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 14:02:14.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e12c1e23c05e3a42136e0196a0f09da87a238468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 15:18:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ae57b1fe1b4d2c898f78980f037879\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/821ffbea1b154d338a4ad3eff238d13d : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d0c55d4d32df4dc1a92cece3969a7136 : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/61122b697d8a4dd7a12da499409ec619 : SUCCESS in 17m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ba24e9f1dd624b2792ad04ab08e99e75 : SUCCESS in 1h 02m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b77db613bd54479ca62153ec2dc4135c : SUCCESS in 1h 05m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a261e559f29f4443b1889069eab4bd9a : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/108a608b38f74b02a8be0ce4c2aa0696 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fa3d134990941908d44c90a4c9888df : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17d38bcd30bf43db8b16abac51358e16 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f03e5758d84738b19fd6cbb801e84a : SUCCESS in 5m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/54e3800589df4d09885530a996a955e7 : SUCCESS in 18m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8a4157a5c0d640cab83551620811a82e : SUCCESS in 1h 09m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/31515dfa3e0c404cb0e7f78a6b2bdd75 : SUCCESS in 1h 07m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f44e8e03b5b54a00a0cd5ed8d726a11e : SUCCESS in 13m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e76d62bb17c40d9a273796b08406eb3 : SUCCESS in 51m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a9d7d0c4abe849ceb9686008cd5a2711 : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":4},{"id":"95de51ff66a7b3cbd023141d4445ffa9b82f6ed1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 15:35:06.000000000","message":"Patch Set 4: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f0664009132313bcf75976f448b6c9ad09468d62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 15:40:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9332b3b429242d3a3e3f13e161d5b0b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/20dc00d1477240e181e8aac30e0b2201 : SUCCESS in 1h 24m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/561ab989d60744f39ea7594d081bb532 : SUCCESS in 1h 33m 38s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3b37e052360444aaa0ecb9c79c509a0b : SUCCESS in 1h 16m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/481698dd92da4187abafba79926bacba : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f35f5e07bdc424ea02790895bdeb74b : SUCCESS in 2m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/660daf9575054f8cbbf44ed690acdaa8 : FAILURE in 1h 36m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2309ad9f0760460189d9b6b792a78c90 : SUCCESS in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb560b0dbbc578c554de171346f187cabef807f4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-23 16:13:18.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d0117e512f5abd0e90dfa8d8dbdd8d725c11eba5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-23 16:31:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c0c3bd04d323f32008a351b13bdcab7db890563c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 18:02:40.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d15e80b77304de85e2e8088508bf7b301a3f72a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 18:03:25.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"daf79bb7a8b1dda1b67d090f588e2ef7f8223300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 18:03:47.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"b7df4a198340accf77550134ae5dd1b038a6977d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 19:47:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c389935fe2cc4299abd1fc425221cb98\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d612325e292e45fb907a79be6d6cab38 : SUCCESS in 23m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a75f65ecc7ba44e2a10f95e6bc8996c9 : SUCCESS in 1h 15m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c9ba7e0dfbca4913870ea54790c23246 : SUCCESS in 16m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/66811efdedd34ecd80dc8f2f01fc240f : SUCCESS in 52m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/77082107e7864d669a9d72e7c1121eaf : SUCCESS in 1h 07m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c6e08bb1d841d986deff6799d3f039 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75f03251e12243ba865b987a7012de4c : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd90b0d67e7647cca4bb97e50732aafe : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38f810e9ae7b485296b7bf780df1bec7 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43645bfc0b545cca89aae0368c22fd4 : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13bdf53276a7497aa5dffd80814acec3 : SUCCESS in 22m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cac88ff4100447fb89e4d9464a0867f4 : SUCCESS in 1h 13m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b74f6e78c22746bba314b3664fb62c49 : SUCCESS in 1h 36m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/76621dfccc5d45feb23cb953b5abdd48 : SUCCESS in 17m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/267f6c0b370b45e8aeb424186fc95227 : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/99990a4df05b42d5bef947f7dd752a9d : SUCCESS in 1h 09m 38s","accounts_in_message":[],"_revision_number":7},{"id":"61bfb46002f340d70fc62790276ca810f08894d5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-02-23 20:07:08.000000000","message":"Patch Set 7: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"689ce1170702d979b2ca7598585f8b3ea177a915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 20:07:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"66374214cb8195cedcad3e3be9dc69d06897e7cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 20:41:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85cbc924f6a143b7b75a6ee7174309c5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/539c2218071748e98bf77267a7b3ec92 : SUCCESS in 19m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/17b79c5b4013420bbf5e925b4054df0a : SUCCESS in 1h 05m 53s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6b9cd915e44c4daebd4c9c4558101411 : SUCCESS in 2h 35m 23s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4e5330aac114f17b7194cd00b115284 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1b37d4a944843f6bef995f59e75a7aa : SUCCESS in 4m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/95f3fe3b71034b549920be30c9724fe2 : FAILURE in 20m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/285553ec901a455ba7e20e8dabcfc150 : SUCCESS in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c23c8f8bf72a5b14d74b29727e01ce4e02659a1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:20:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ba0eb80cc84f7e9056756f8e5bfd5a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c1638acee0474d6db7da96e4793184d5 : SUCCESS in 14m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/07b9a51f1aca4c1d9f51374782f1a6bc : SUCCESS in 56m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/04835ebe2c874978adcdfdf1a34248c4 : SUCCESS in 1h 06m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c060b5b4eb4015a080d5c2280b4682 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/422ace8d43a44fbaaae6cebbe8d22814 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/664ae4113c704d598746fd5b7600b706 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03f1cdbe55e41d7838ab5ee4f210f48 : SUCCESS in 4m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3970bde821354b6abd6fa5d8c34f7477 : SUCCESS in 18m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/14f0af4c761b4604bdd6d4353d69d2f0 : SUCCESS in 1h 07m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2757a719700e48b2952ad43d2563039c : SUCCESS in 55m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/90b20184d9fd4f1897c1c4b790c77371 : SUCCESS in 13m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5477e5796a1744d58f4116696a9beed4 : SUCCESS in 46m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cb4d4b11e2404ba2aa6d66cb16a0dad2 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":7},{"id":"ac14b1cfcfff7ee55d4f4df1d8b05032730b2562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:20:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8391ce94c9ed5a310201c1324ea4e4fb8117a705","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:21:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15eec9c4d71c45a8a2dea19834d5f06e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90485fa1426d484fa8ffec888a949828 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7},{"id":"48358bbac08baf1af8de346335f216aed84c030c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-24 07:21:10.000000000","message":"Patch Set 7: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a83585d9286f007c785498713d48ca61edcdd95f","revisions":{"459d92da3e35c5e8f13f3cb92df93977d4468ca8":{"kind":"REWORK","_number":1,"created":"2023-02-22 07:32:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/1"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:43.000000000","tz":60},"subject":"toolbox: uncap openstacksdk and use openstack.cloud 2.x","message":"toolbox: uncap openstacksdk and use openstack.cloud 2.x\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/459d92da3e35c5e8f13f3cb92df93977d4468ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/459d92da3e35c5e8f13f3cb92df93977d4468ca8"}]},"branch":"refs/heads/master"},"0181235a0cc0224754ec10a5efbf1e3c0452001c":{"kind":"REWORK","_number":2,"created":"2023-02-22 09:33:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/2"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 09:33:38.000000000","tz":0},"subject":"toolbox: uncap openstacksdk (install 0.103)","message":"toolbox: uncap openstacksdk (install 0.103)\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0181235a0cc0224754ec10a5efbf1e3c0452001c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0181235a0cc0224754ec10a5efbf1e3c0452001c"}]},"branch":"refs/heads/master"},"abda27aaa6a79db412068dfa81be05f8a7e3bf6f":{"kind":"REWORK","_number":3,"created":"2023-02-23 14:01:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/3"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 14:01:09.000000000","tz":0},"subject":"toolbox: uncap openstacksdk (install 0.103)","message":"toolbox: uncap openstacksdk (install 0.103)\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abda27aaa6a79db412068dfa81be05f8a7e3bf6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abda27aaa6a79db412068dfa81be05f8a7e3bf6f"}]},"branch":"refs/heads/master"},"a88771c30744eb5a5bb5cb5f74a3191089e182aa":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-02-23 14:02:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/4"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 14:02:14.000000000","tz":0},"subject":"toolbox: use ansible-collections-openstack \u003e2","message":"toolbox: use ansible-collections-openstack \u003e2\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a88771c30744eb5a5bb5cb5f74a3191089e182aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a88771c30744eb5a5bb5cb5f74a3191089e182aa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3d4ea2f76fb56a32abbff8e9a652173f757a7553":{"kind":"REWORK","_number":5,"created":"2023-02-23 18:02:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/5"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 18:02:40.000000000","tz":0},"subject":"toolbox: use ansible-collections-openstack \u003e2","message":"toolbox: use ansible-collections-openstack \u003e2\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d4ea2f76fb56a32abbff8e9a652173f757a7553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d4ea2f76fb56a32abbff8e9a652173f757a7553"}]},"branch":"refs/heads/master"},"8b576e18e6efff7ae507613528c2e6d6582de96b":{"kind":"REWORK","_number":6,"created":"2023-02-23 18:03:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/6"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 18:03:24.000000000","tz":0},"subject":"toolbox: use ansible-collections-openstack \u003e2","message":"toolbox: use ansible-collections-openstack \u003e2\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b576e18e6efff7ae507613528c2e6d6582de96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b576e18e6efff7ae507613528c2e6d6582de96b"}]},"branch":"refs/heads/master"},"a83585d9286f007c785498713d48ca61edcdd95f":{"kind":"REWORK","_number":7,"created":"2023-02-23 18:03:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/874725/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/874725/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/874725/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/874725/7"}}},"commit":{"parents":[{"commit":"7bbf36cca0f77ebaa48b123f7f226a64065b74ff","subject":"Merge \"CI: Additional disk followup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbf36cca0f77ebaa48b123f7f226a64065b74ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:32:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 18:03:47.000000000","tz":0},"subject":"toolbox: use ansible-collections-openstack \u003e2","message":"toolbox: use ansible-collections-openstack \u003e2\n\nChange-Id: I6564886f68a997b82b49b6f975283a6d95831ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83585d9286f007c785498713d48ca61edcdd95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83585d9286f007c785498713d48ca61edcdd95f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
