)]}'
{"id":"openstack%2Fkolla~874730","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41","subject":"CI: Retry docker push on publish jobs","status":"MERGED","created":"2023-02-22 07:38:17.000000000","updated":"2023-02-22 13:24:12.000000000","submitted":"2023-02-22 13:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874730","cherry_pick_of_change":874055,"cherry_pick_of_patch_set":1,"meta_rev_id":"ecbdf10a0eb505bd84f4383ceea48693f0fa0710","_number":874730,"virtual_id_number":874730,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 13:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-22 09:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-02-22 09:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 10:49:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-22 10:56:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65e65b3b2b13a670f635138328886a149a3d116d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 07:38:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"208bac0052f9193d94b7158c0b37ca9f9b63efe2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 09:07:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fa461be48006a791f6b7f363440bf56431e0c56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:49:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2a9beb18de4a82a2aeac1eb70939d3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/907f36b1017346a9966ae42ed71d53ea : SUCCESS in 48m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb9c346ab99248c192c9cd96d6a89ded : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebf719364b5047c4b3e0c52c3dd9ce82 : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf16d43ee6ec4cd18f5c9acf364952e0 : SUCCESS in 2h 03m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b75a9c8aa8d789867a7f6b13fe3ca1aa7b8dc08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:56:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/029d4af978ac471fba6aef7c1e618bac\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/51fdfa863cfb41e5b21f9138a0de25e5 : TIMED_OUT in 3h 01m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9a4bd73442ee4f4ebb3b1c628913a174 : SUCCESS in 31m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/28d4c53bc3734926b27b02e85b03cbc4 : SUCCESS in 1h 16m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a7368731012411da80d5151e5986e19 : SUCCESS in 1h 08m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/39df8cd06903416d86ca1f083f6a5b1e : SUCCESS in 1h 13m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3c2ba9a53c664638ab3d814554995337 : SUCCESS in 24m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9b6ec03a7cb84d7495eeeab67859863c : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/98ba30a8ed4b494eade7dad990c7fc16 : SUCCESS in 1h 00m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b803d709172f4d16a4683ecded4d26e4 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beff6e60709c4845bedfb18e92b83fd4 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f74928b045438db358cd882c0e78dc : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61716e530ae1483a833a0d3c0f91f67c : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae24bde7ca044df9027ae19939501ef : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0df5a32cc9194fad9ce5eb2c65796287 : SUCCESS in 19m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b71bf961e5b74345a06eb5ded0202592 : SUCCESS in 23m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39a830fab96f45c88740fd0ace8f1978 : SUCCESS in 51m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af89bb39ddde44658d0d4fe3dee0e68d : SUCCESS in 57m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c91483a6e0404a61bc2a9ab5d4d16370 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"666e6c0ec7bfee757c723ba6fc4fb4f861779221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:57:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e68e1ec62d6ad5082624b20a198703867dc9e58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:23:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b28bdb0a9f349cda2277a68017c2124\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0d303c6d72b42df8096e646db7b1b09 : SUCCESS in 33m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8ff6429563d04d7eaf3d334944f1dc90 : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d55e5bf57cea413eb98f72a035117088 : SUCCESS in 1h 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78acc26dbb644cdfbf02fce855c64c23 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7cbc7c2e1ec46108ff6c1204b6bda45 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d6b3b674b8f41dcaf1de51e918bedba : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f551e163daf42c4b1bc4debc45f16e8 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba763e60593c4bca978278a2f848e972 : SUCCESS in 25m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1730a1e6e594f3886dabb54495a331b : SUCCESS in 1h 06m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e82bfa1feea3493d859368f0c923b5b8 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"679e78f9261044ff88484136ecbdb0bee1a6722c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:23:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ecbdf10a0eb505bd84f4383ceea48693f0fa0710","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:24:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbfff59c988d4194bb0853b9d8343631\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46d7319ed254c31a8ff35c4d978543b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a7f1b2fe6b554b005c0b1b9e711a64353e61b85","revisions":{"3a7f1b2fe6b554b005c0b1b9e711a64353e61b85":{"kind":"REWORK","_number":1,"created":"2023-02-22 07:38:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/30/874730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/874730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/874730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/874730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/874730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/874730/1"}}},"commit":{"parents":[{"commit":"1691f36a329e37cba041f367037f8fae1880592d","subject":"CI: store only templated out Dockerfiles in work_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1691f36a329e37cba041f367037f8fae1880592d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-16 07:23:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:38:17.000000000","tz":0},"subject":"CI: Retry docker push on publish jobs","message":"CI: Retry docker push on publish jobs\n\nWe have a high number of POST_FAILURE jobs failing on at least\none container push - let\u0027s fix that.\n\nChange-Id: I0015dd4d3ffe5b3e1202ebcfde2b2c72c4655b41\n(cherry picked from commit 0887d2e789e8fcd41b49c6346034c4774ea919dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a7f1b2fe6b554b005c0b1b9e711a64353e61b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a7f1b2fe6b554b005c0b1b9e711a64353e61b85"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
