)]}'
{"id":"openstack%2Fkolla~874731","triplet_id":"openstack%2Fkolla~master~Id409c2b5423791d4ca9626e376de07ef7c1a8d8e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-02-24 07:32:36.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-02 13:49:58.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-02 13:39:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Id409c2b5423791d4ca9626e376de07ef7c1a8d8e","subject":"Revert \"bifrost: mark unbuidable for Ubuntu\"","status":"MERGED","created":"2023-02-22 07:53:58.000000000","updated":"2023-03-02 15:24:34.000000000","submitted":"2023-03-02 15:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"revert_of":874383,"submission_id":"874731","meta_rev_id":"0be86b515969a9f4662e86e3017dc9757638100b","_number":874731,"virtual_id_number":874731,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 15:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-02 13:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-02 13:42:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-02 13:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-02 13:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 07:53:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-02-22 07:53:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 11:48:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-03-02 13:39:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-02 13:42:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3c08510040763de7a669f16b13dceee42dd38c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 07:53:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3008f1ce7c77ce057510b15908c9b10777496c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:31:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d77585452b403493ab575fcb9b25dd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8c9a783689d042e48a72f1e96015a4f6 : FAILURE in 24m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/06f4c24f53c1493dbfd63edb07d42d3b : FAILURE in 21m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4beedb1eed034d87802cff1ed7f40e54 : FAILURE in 18m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/826fc37579974aa59743ab4923b8a07b : FAILURE in 18m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2c12bb2daa284566aa763250b5d54f59 : FAILURE in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30fcd84a4648417d8e6e9fdd2eebb159 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c3441c4e10a441e99c05904cf42f9e3 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ee1f04c42f34673b7bae2f314c96cb6 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab5d6d476eb344a0aa7bf0aec16b10a8 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9203b26fa93546299792fddbcc90ed82 : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4041c058c5454341a2f035764f1df04e : FAILURE in 19m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/782399627edd47618b258eb1c17140b4 : FAILURE in 21m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/18ffc7701f6249a28a64b95c2c1946b9 : FAILURE in 20m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f9b18c1b0234b2fa3b2a7f4f8dce0cf : FAILURE in 25m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/efe4b63b95e44b5ebc7d700707292ea8 : FAILURE in 12m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7f6b2bda13694310a77827342de9cc15 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2fd28f9f6760ce4455bce36dad237bc66e2c05c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 11:18:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464c581200654f16bce7fa27f07c4aef\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1a27aa828f2c4c2a913277ca59a8fd7d : FAILURE in 1h 39m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/98a528808902408099170561dc72c7bc : FAILURE in 33m 27s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8bcccccc7a8247159fba63caa2c6848f : FAILURE in 58m 00s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03c09d8082ae4b538d2be7545453bfb9 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94221fbabde4446085513d2feda593a7 : SUCCESS in 3m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cd2bda04493b451887aa9b8c0ac93902 : FAILURE in 1h 47m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/83faf5a7314d4bb580a862ba391ee9a1 : FAILURE in 2h 00m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63315b5d8a6785355fed4db0327995e83201db5e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-22 11:48:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4558b2783282b4d809df4094b896a19f6b92f3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-24 07:36:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7beec0bf50fbacad106a8cdf0ef2ad0bc89ea48f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-24 07:36:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99eb2c10863ab72bd4d633a78590f3e273a31eb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-24 09:09:23.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e148c8c388d18e998bbda6bb7a83e6ea58030310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 10:35:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f0b365f2efc4e6e928a55873fcfbba5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/712eaa38e9244cac86b387bb4b4218c5 : SUCCESS in 16m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/594a0b7004ae40d697c8519b509b8730 : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83b3aa3eb96044fabdd8b0cfc9468cec : SUCCESS in 15m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0deb8932a87e4d8880e562324351ee66 : SUCCESS in 1h 02m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/538d07b90077464689612ba9d8a50d71 : SUCCESS in 1h 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8714ac0d21c64782a0426266667737da : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/851b9a9696af4b2a9dc776d754037305 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6df5673de8e540089c4fa020772af3b7 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f4553c55f1a40b8a8ac80d9ae70e91c : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb875c4725c442ea8d5e296c6cd722b5 : SUCCESS in 7m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dc73cac5182c444a87ff5fe8822e63d7 : SUCCESS in 27m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/855b8abd3bac46319c38b3473f672edf : SUCCESS in 1h 15m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3746176bd0694cbd851378a284c4e798 : SUCCESS in 1h 17m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/86731156809a40d38b28f4b21b355664 : SUCCESS in 20m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f3899adaf1c14169bf36415658f55833 : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b9271f4ce0414d3b9066ba9448d7b312 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":3},{"id":"a596188d2f81efe25ed02b61d6bead666b585902","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 10:46:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7a2e7bc1dfd4eebba37aa1838265d8f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1d0683e43fe0453b981ea15e67b74f46 : SUCCESS in 24m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d38442eaec64b3f940d32a8bbaa0969 : SUCCESS in 26m 03s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25d8bd91487a409787087fba3cac518e : SUCCESS in 1h 35m 15s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1899e1f02b3f4086981f8db300660f0a : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a60caba99f248ee8038c69b81f00186 : SUCCESS in 3m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dee491ad5c134f0e917fb2c2e700441d : FAILURE in 24m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9ccaaeea00be41768c72d0337a4d3347 : SUCCESS in 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30f8c724457b83d6b4f8e2be3d75af42f0d7645b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-02 13:39:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6ef4a108c50913c30a84478c106024e53cf563e9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-02 13:40:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8fda1a2a50853b5d7aeefbcd4b56adc14576b4d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-02 13:42:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"371d7803a8cfea8e88c51ae5d7fb60ef7ec37371","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-02 13:49:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c2c3a5aeb7cba46029e5a2bd8016eeb3f60053b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 13:51:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"96b73a5233de7121e0c4de662b9aae701570e793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:23:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e424b4c75a0642feaf8bca7e39b1439f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0f3483bba228434fbe69e92c41975ed1 : SUCCESS in 18m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4d705d7ecbbc4f1385407bd50b72f1f9 : SUCCESS in 1h 04m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2142a39f5ba94dc6900eda7f74cb8915 : SUCCESS in 1h 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5775cfd76f345b6925cb680081e04ef : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60e5219ac0546898e67262c3e99b1b3 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/878754740c1e475badf7654824da08ec : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6416ac9328bd4beaaa2c0f87b7a9976f : SUCCESS in 5m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/edbe0a41a6f1479a915d86150ec6860a : SUCCESS in 18m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/44f053ddc22347cf8ad0627819a9a67c : SUCCESS in 1h 10m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/023074cdef124caeba6a3932fb5f4893 : SUCCESS in 1h 22m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c3825e2828fc41f8bc0c9870efa8d443 : SUCCESS in 19m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b459d2be060a4a8c95fdf44ab44e4f4e : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a5ea8435d3e74d1ea679cbfa3b15308a : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c95f0902f4c41cf89d24460e4dffc1712a637cc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:23:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0be86b515969a9f4662e86e3017dc9757638100b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:24:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a8f395d7114eb58a1943d639df752c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152d2a289b6e4387aee3cdbe6c71c097 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"48d44c286fee004877b9c6240b015488ea4f7d02","revisions":{"d37cf815cf6f1c2bc266683c7591edc0926c8a13":{"kind":"REWORK","_number":1,"created":"2023-02-22 07:53:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/874731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/874731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/874731/1"}}},"commit":{"parents":[{"commit":"b2f747fbba7adf3059d0da9bd899788b4a45763c","subject":"bifrost: mark unbuidable for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f747fbba7adf3059d0da9bd899788b4a45763c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:53:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:53:58.000000000","tz":0},"subject":"Revert \"bifrost: mark unbuidable for Ubuntu\"","message":"Revert \"bifrost: mark unbuidable for Ubuntu\"\n\nThis reverts commit b2f747fbba7adf3059d0da9bd899788b4a45763c.\n\nReason for revert: Fixing in bifrost\n\nChange-Id: Id409c2b5423791d4ca9626e376de07ef7c1a8d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d37cf815cf6f1c2bc266683c7591edc0926c8a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d37cf815cf6f1c2bc266683c7591edc0926c8a13"}]},"branch":"refs/heads/master"},"d6cd47256248b5709afde2f2796fe5d0f8b1fb99":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-24 07:36:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/874731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/874731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/874731/2"}}},"commit":{"parents":[{"commit":"b2f747fbba7adf3059d0da9bd899788b4a45763c","subject":"bifrost: mark unbuidable for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f747fbba7adf3059d0da9bd899788b4a45763c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:53:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:36:19.000000000","tz":0},"subject":"Revert \"bifrost: mark unbuidable for Ubuntu\"","message":"Revert \"bifrost: mark unbuidable for Ubuntu\"\n\nThis reverts commit b2f747fbba7adf3059d0da9bd899788b4a45763c.\n\nReason for revert: Fixed in bifrost\n\nhttps://review.opendev.org/c/openstack/bifrost/+/874650\n\nChange-Id: Id409c2b5423791d4ca9626e376de07ef7c1a8d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6cd47256248b5709afde2f2796fe5d0f8b1fb99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6cd47256248b5709afde2f2796fe5d0f8b1fb99"}]},"branch":"refs/heads/master","description":"Edit commit message"},"48d44c286fee004877b9c6240b015488ea4f7d02":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-24 09:09:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/874731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/874731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/874731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/874731/3"}}},"commit":{"parents":[{"commit":"b2f747fbba7adf3059d0da9bd899788b4a45763c","subject":"bifrost: mark unbuidable for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f747fbba7adf3059d0da9bd899788b4a45763c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:53:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 09:09:23.000000000","tz":0},"subject":"Revert \"bifrost: mark unbuidable for Ubuntu\"","message":"Revert \"bifrost: mark unbuidable for Ubuntu\"\n\nThis reverts commit b2f747fbba7adf3059d0da9bd899788b4a45763c.\n\nReason for revert: Fixed in bifrost\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/874650\n\nChange-Id: Id409c2b5423791d4ca9626e376de07ef7c1a8d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48d44c286fee004877b9c6240b015488ea4f7d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48d44c286fee004877b9c6240b015488ea4f7d02"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
