)]}'
{"id":"openstack%2Fkolla~874914","triplet_id":"openstack%2Fkolla~master~I3b90d1e9e083128af60b03bdf1e2f83fb66fceb3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-23 15:20:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3b90d1e9e083128af60b03bdf1e2f83fb66fceb3","subject":"toolbox: Stop using u-c","status":"ABANDONED","created":"2023-02-23 11:54:11.000000000","updated":"2023-02-23 15:20:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0373793a01f132571894fccd614c4551eef0d3a","_number":874914,"virtual_id_number":874914,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-23 15:14:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 12:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e3dc201425f3694107fd8c42ad4ad880becd0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 11:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1e4f4e87099715118c5ebdfd1261a50e3f36a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 12:54:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fe1d82abfd742a09c0969f49bb1958a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/66fe1ec782364ec4a1bed31381697fc2 : FAILURE in 19m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7c64b8276a54402d8e20da353944554c : FAILURE in 21m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/224cca8870714bc1b3e1b448a806a46e : FAILURE in 18m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64cfc70935a443c288628972f67a273a : FAILURE in 26m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/efd0324ac8404230ac3d563538509be2 : FAILURE in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e44b01e872e486ba537f8b1ea52c336 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918033fb78664297b1c37c20560fd563 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b8ab230127141eb8b1b2025e8a848e8 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61586295dac543b5824aa4ae9458308b : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1b11b5c5464551bdca18aca58ce46f : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ac6d44c5603649fc8ff1b72771353a51 : FAILURE in 19m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/91c1c36f83dc48778c4ce8106fb07a07 : POST_FAILURE in 37m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cdc9d408c17a444481460b406b00fbb0 : FAILURE in 19m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d6c9b291e59449e0ba3de84105643b3a : FAILURE in 19m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/21ff87111e0a442d90a3c2fd3ab4d11e : FAILURE in 15m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5c5dccc230cf4930bbbab7f529d9a89f : FAILURE in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a52a81ea80a618fa02c96508ce57ae71891faf4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 13:14:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb55285a857d45e78f4fc78e927e571f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cd51117a18f243cf96515677630c2a29 : FAILURE in 1h 13m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6eb1f1b064d64531a665b0c8d89c00e2 : FAILURE in 1h 14m 58s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/203f6ed72c574bc0b4e1387b2275f826 : FAILURE in 55m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6480ea0a5dfc46b08d04fff361493b51 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d7ef80220db4bf484641cb16908fc20 : SUCCESS in 4m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7ad41758a80e4dbcb21565248e83edf8 : FAILURE in 1h 16m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bbd7c696030a43c98ebd0f9c1e4958cf : FAILURE in 1h 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb27684a337ed32d01582d40199aec98071614b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 13:15:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9adfa0f414fc7a3218982bc975da5688776d86d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 13:28:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e9fd1430dc8cf2253b68eac98d1ecd987e63b55f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 14:52:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d4b0a2142b47caa668a579df8161b1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/09b2e02ea4b24b1f97c24f188370c15b : SUCCESS in 28m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c674a01509b646efbfcaca2f1cba9f32 : SUCCESS in 22m 02s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/344747dde801404e894046eacda81eff : SUCCESS in 1h 22m 14s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/317a728063d24a5a95d4b57febcaafdf : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/77d81ff7936440b0ab8d5fe45fe421eb : SUCCESS in 2m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f7dfac8d28ab480ebfb43d44157bdea0 : FAILURE in 22m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b1812280c4e545cca4a0b9416f9d66b4 : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1a6f9bfc761341e20933268d3480fa3e195d1ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 15:14:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19da4d9ad2044528b87a558a95a9ba5a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/34c94d380e5844ada785e0bda0fc4fe2 : SUCCESS in 25m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e6b194d13935474d9b5a159e8d855b6e : SUCCESS in 1h 40m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42146e9e7d924b12ac39048f469a7a58 : SUCCESS in 21m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3cced23edde848168f5bcfb5931c5177 : SUCCESS in 57m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/185cca16bc504c06a297864a5e98ab21 : SUCCESS in 1h 07m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f292bbc83fb4e3184128bec8a7a8eec : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02853897220349d2b5ccbfa98b452f27 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb6c5357f3ac4c07a5333e41270c3485 : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6ec80b788424b87863c858b489abebf : POST_FAILURE in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/352be8b367da487d9b8d74bf4ee7e782 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/adae18a573df4eb18409d3b951949e79 : SUCCESS in 20m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9a89af93c35f4a368d3125170c3ef603 : SUCCESS in 1h 03m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bd7392dace994891ba54b1a60f0ba291 : SUCCESS in 1h 13m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/540705b4c6a54c50b9cfcb6c4edfbce0 : SUCCESS in 18m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b991ba2ce1f847e5a3c25e3386a18992 : SUCCESS in 48m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0c776acc0fd04d6aad3a62d8009f6862 : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f0373793a01f132571894fccd614c4551eef0d3a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 15:20:58.000000000","message":"Abandoned\n\nIn favour of https://review.opendev.org/c/openstack/kolla/+/874725","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f5e353f9d6cf9746f7dfa340aee6ba109fbd59b8","revisions":{"b953d37dea73a09460936acef0de49339ba4aa09":{"kind":"REWORK","_number":1,"created":"2023-02-23 11:54:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/14/874914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/874914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/874914/1"}}},"commit":{"parents":[{"commit":"83cb88fdad9433418a2d45cb923de167b2135106","subject":"Merge \"[Doc] Add venus project in README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83cb88fdad9433418a2d45cb923de167b2135106"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 11:52:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 11:54:09.000000000","tz":60},"subject":"toolbox: Remove dogpile.cache and pbr from u-c","message":"toolbox: Remove dogpile.cache and pbr from u-c\n\nBuilds are breaking now with:\nINFO:kolla.common.utils.kolla-toolbox:The conflict is caused by:\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.22.0 depends on dogpile.cache\u003c0.7.0 and \u003e\u003d0.6.2\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.21.0 depends on dogpile.cache\u003c0.7.0 and \u003e\u003d0.6.2\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.2 depends on pbr\u003c2.0 and \u003e\u003d1.4\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.1 depends on pbr\u003c2.0 and \u003e\u003d1.3\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.0 depends on pbr\u003c2.0 and \u003e\u003d1.3\nINFO:kolla.common.utils.kolla-toolbox:    The user requested (constraint) pbr\u003d\u003d\u003d5.11.1\nINFO:kolla.common.utils.kolla-toolbox:    The user requested (constraint) dogpile-cache\u003d\u003d\u003d1.1.8\n\nCaused by u-c pinning to openstacksdk 0.103 (and pbr/dogpile.cache\nto similar versions, which are not working with openstacksdk 0.62)\n\nChange-Id: I3b90d1e9e083128af60b03bdf1e2f83fb66fceb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b953d37dea73a09460936acef0de49339ba4aa09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b953d37dea73a09460936acef0de49339ba4aa09"}]},"branch":"refs/heads/master"},"b8d0075888872abb878184628557e6778105832e":{"kind":"REWORK","_number":2,"created":"2023-02-23 13:15:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/14/874914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/874914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/874914/2"}}},"commit":{"parents":[{"commit":"83cb88fdad9433418a2d45cb923de167b2135106","subject":"Merge \"[Doc] Add venus project in README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83cb88fdad9433418a2d45cb923de167b2135106"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 11:52:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 13:15:47.000000000","tz":0},"subject":"toolbox: Remove dogpile.cache and pbr from u-c","message":"toolbox: Remove dogpile.cache and pbr from u-c\n\nBuilds are breaking now with:\nINFO:kolla.common.utils.kolla-toolbox:The conflict is caused by:\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.22.0 depends on dogpile.cache\u003c0.7.0 and \u003e\u003d0.6.2\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.21.0 depends on dogpile.cache\u003c0.7.0 and \u003e\u003d0.6.2\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.2 depends on pbr\u003c2.0 and \u003e\u003d1.4\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.1 depends on pbr\u003c2.0 and \u003e\u003d1.3\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.0 depends on pbr\u003c2.0 and \u003e\u003d1.3\nINFO:kolla.common.utils.kolla-toolbox:    The user requested (constraint) pbr\u003d\u003d\u003d5.11.1\nINFO:kolla.common.utils.kolla-toolbox:    The user requested (constraint) dogpile-cache\u003d\u003d\u003d1.1.8\n\nCaused by u-c pinning to openstacksdk 0.103 (and pbr/dogpile.cache\nto similar versions, which are not working with openstacksdk 0.62)\n\nChange-Id: I3b90d1e9e083128af60b03bdf1e2f83fb66fceb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8d0075888872abb878184628557e6778105832e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8d0075888872abb878184628557e6778105832e"}]},"branch":"refs/heads/master"},"f5e353f9d6cf9746f7dfa340aee6ba109fbd59b8":{"kind":"REWORK","_number":3,"created":"2023-02-23 13:28:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/14/874914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/874914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/874914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/874914/3"}}},"commit":{"parents":[{"commit":"83cb88fdad9433418a2d45cb923de167b2135106","subject":"Merge \"[Doc] Add venus project in README.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83cb88fdad9433418a2d45cb923de167b2135106"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 11:52:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 13:28:57.000000000","tz":0},"subject":"toolbox: Stop using u-c","message":"toolbox: Stop using u-c\n\nBuilds are breaking now with:\nINFO:kolla.common.utils.kolla-toolbox:The conflict is caused by:\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.22.0 depends on dogpile.cache\u003c0.7.0 and \u003e\u003d0.6.2\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.21.0 depends on dogpile.cache\u003c0.7.0 and \u003e\u003d0.6.2\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.2 depends on pbr\u003c2.0 and \u003e\u003d1.4\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.1 depends on pbr\u003c2.0 and \u003e\u003d1.3\nINFO:kolla.common.utils.kolla-toolbox:    openstacksdk 0.6.0 depends on pbr\u003c2.0 and \u003e\u003d1.3\nINFO:kolla.common.utils.kolla-toolbox:    The user requested (constraint) pbr\u003d\u003d\u003d5.11.1\nINFO:kolla.common.utils.kolla-toolbox:    The user requested (constraint) dogpile-cache\u003d\u003d\u003d1.1.8\n\nCaused by u-c pinning to openstacksdk 0.103 (and pbr/dogpile.cache\nand os-client-config to similar versions, which are not working with\nopenstacksdk 0.62 - the list seems to be growing over time)\n\nChange-Id: I3b90d1e9e083128af60b03bdf1e2f83fb66fceb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5e353f9d6cf9746f7dfa340aee6ba109fbd59b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5e353f9d6cf9746f7dfa340aee6ba109fbd59b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
