)]}'
{"id":"openstack%2Fkolla~874916","triplet_id":"openstack%2Fkolla~master~Id20394d06747b7f56e8745f129149c3f1c492ca7","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-18 05:17:02.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Id20394d06747b7f56e8745f129149c3f1c492ca7","subject":"toolbox: use venv instead of virtualenv","status":"MERGED","created":"2023-02-23 12:09:10.000000000","updated":"2023-05-18 06:38:12.000000000","submitted":"2023-05-18 06:37:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874916","meta_rev_id":"ce2315f9c6480bfa1fac4cd9e7b1b7316557e71e","_number":874916,"virtual_id_number":874916,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-18 06:37:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-18 05:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-17 14:10:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 15:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-18 05:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-17 14:02:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 12:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-23 13:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 14:10:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-17 15:29:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"ba5863f809f4f292b00f44f9ec6658987915c378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 12:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f5469efdf2f6c6193b402d441ce1ad5b2318e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 12:29:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0e59c3fc890823804dbc74be61f1d5141b50b552","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 12:58:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d7cbadb389d4ea28537065f8cb8ff47\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/47f2ed553b8f49c59d0d3b3eb6188eb2 : FAILURE in 22m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/16ac80ef3bee4b10afa9fd655b23b9bb : FAILURE in 17m 34s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2751fb71feb9452394fe89fa55e70fcd : FAILURE in 27m 14s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec3e85b2f3774d4391ce034acee72c9b : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b897000afb8f4be19732b7843364dd81 : SUCCESS in 3m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bb7e5e58442143cd85687190efd9f4e0 : FAILURE in 19m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b1b6853c5e2848578fc0d61a56e8f17f : FAILURE in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57338b99e8f5e5a90f9c966154f711ab7d08424a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 13:06:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/565ad06b2878477eb79eb5de9cd493a1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/35fa78d970fe472d83638d0d99f6fa41 : FAILURE in 30m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/352e88ea671f49f18ba83a8971975104 : FAILURE in 25m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a7b0f21e3d154e68a5f6771c543779b0 : FAILURE in 19m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8b3f916dd1b54fa59211952e41e72cd6 : FAILURE in 20m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/599ab0892f994f1a983d65fef688b815 : FAILURE in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69711a20c8224504ae1cb764982cf8b8 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92e75624a7714bfb9678e82037e227b5 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50a77916fd164f1bbb67eef11ddf3e5b : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fd866b7346d48e9b58ed622c573c9f7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3628f9284484fe38e7104f6df0c520a : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55538d6324724142b2637f307fbb580b : FAILURE in 16m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4657c91426d141ae8fe102eaa1fdf289 : FAILURE in 21m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/65e0514839064ec49f61734138dcf8d5 : FAILURE in 19m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/544492674f0742808e901b47f6c02004 : FAILURE in 14m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa0dc97a9d274b0fbaae5c057a811834 : FAILURE in 19m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b5cf3c9a55dc4d659b1426f1ccbe0b4a : FAILURE in 15m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5d71915bb4f26749fbfec1368cb7c5b8101c9137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-24 07:32:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9278c73335591ebfaf9266202b4c57ba65876b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 08:01:43.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5da8708ce44c33a32b4cf4906302eb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8bf73ad4d67a4ad19d8a79c4fcadfc68 : SUCCESS in 25m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e78239cbbd1549eda1385dec7e2f7540 : FAILURE in 22m 42s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afd50b46059944079c382bd25a9a703f : FAILURE in 24m 39s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6803c83179a943aebbb9ea14c0e98fbf : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c22c82768a242ad838e68913aaa7135 : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f187e0930b864bda8fd4d90fdf1b79c3 : FAILURE in 28m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/682d8781f80b467db61ac7cbd29275e0 : FAILURE in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b3394b6ef22bfafee97182eebdbebbf3e6653d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-24 08:03:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7e3cc6d847e0957337097eff81551314a2661be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 09:44:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1a791343e84411b85b1d5443d375558\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3e1b0689dab8450d9434d12cba7611bb : SUCCESS in 24m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a57fc09372044ad2bf15a036e48a8b45 : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/944c0bc6954a4765baf9c6118caab5fb : SUCCESS in 15m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/45c40a527d384859ac7fa09f3ed398fd : SUCCESS in 1h 00m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c545e557ae954a0c8fd29694980670ed : SUCCESS in 1h 04m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e303ce82c4b4a66af9e8e28785b7120 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa123a4bb96d4bb385c76ff021c105ae : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f9cd503c33f46dbabd2444873684cba : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a83daa220b14d5cb302348de8b40722 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33c098589de4271ad1dcefe40587900 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e88bb8ad9124c7c8d01b46ec497f239 : SUCCESS in 21m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9285c56ec9334b2ebf123becc2594de5 : SUCCESS in 1h 12m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b12c07980d9a4e4e8ac1e817cd70b525 : SUCCESS in 1h 17m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c2b37ac04d7541d3bcaa8ffd916a6b50 : SUCCESS in 15m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/543e6c7b4325466e820b60c0813f2f89 : SUCCESS in 48m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/45588f7ab2754b1da102c6b4ffefe8af : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":4},{"id":"f207f5b5a46fa92d15e9e16b27deb0d7a7e5c2f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 10:11:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/164334cac0c04737b3327d083fb2065e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b69b8913e1894c8db34faa9266839a09 : SUCCESS in 22m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6954ef4b206746c4b3a118d11191939b : SUCCESS in 23m 22s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f428162de7245cd925cdb937c891108 : SUCCESS in 1h 28m 06s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6edc6265ccd4223941a4ef611718617 : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/87bf1c91039b46aa8bf9c17ff7e4f64d : SUCCESS in 2m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cb526cca2fb9495c935b7bd5136caeff : FAILURE in 23m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4eefb18c0d074950977a4f461126895f : SUCCESS in 2h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"61e92c3831c880525b4bed15d0347e5ca97ebed3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:58:49.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* docker/kolla-toolbox/Dockerfile.j2\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2fa7762a268474e98a3e7a503d8ce3b79f9587ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:59:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"0f4200ad8091c21d7ebfefee174724beb569deeb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:59:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"1460bf0698612d57dd7a8e5206074996f133168e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 14:00:38.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"795a0d2fc88102ff2f31940c648d7bf39fd12a76","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 14:02:51.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"1708104edf7fa5ef61443235f553a353a88e1c39","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-17 14:10:52.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"25d67fcb1636546d1ead2efffe2529f18204f607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 15:18:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cea4edbf1654098a1682cea889b0d69\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d9ab6778ed7545919f54ad31e83ea7cf : SUCCESS in 22m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a53d6da0f7ce4a4bb30918d3ab50c05a : SUCCESS in 1h 13m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6df3a21f92394ead9e0f422ed8723d92 : SUCCESS in 20m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cc3ae1d0dc5a49a9b0ce599bf22a1b0f : SUCCESS in 1h 11m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a41a4af6c3c476698c42b4d631513c7 : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537413d1afeb41a0badb57d024b01334 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920ced5e1f82449eac36e0bcdab948a8 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea06b1ada54d47f895e3b07c1164609e : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb8a941cf4994fcd931c2cc2d768528c : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb79a71c3e91461784271bea91090be9 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e00fef9457494fa77d1f192c6f8522 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/42648f589de44c17802f704128744e15 : SUCCESS in 21m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8aaefc0bded24c85aad424526febc01d : SUCCESS in 1h 11m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2a080ee9fa434361bfc5ae6a6a2ec96c : SUCCESS in 1h 15m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4aee0823aae74b4db4b5221edb944c1c : SUCCESS in 26m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b528e2cef6414a75b141446df0a3c0b7 : SUCCESS in 53m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f0717eb97bc94dd99b650a7a262a5788 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":7},{"id":"6f6ab958c69ce643488274099dde9018abaa5c39","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-17 15:29:11.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"98e26f2ca45abec6b289245dbd8b0fb27d7e0869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 15:49:43.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53552d6b5aef4e638c2ffa6c065e79a1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a6d295ebb7254607ade2a057748867dc : NODE_FAILURE Node request 300-0021233733 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/88cd8c2cbf194efd9878afbf4a2e7640 : SUCCESS in 1h 45m 11s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36ca451cda3f4d839e5939946cc61d70 : FAILURE in 1h 25m 07s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a640a82f1df7426c922e55cbcdf52a31 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e46ec8149a343e589497d675dd2f9b5 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c0f23b3f33341de96b18d6f4a87ce84 : NODE_FAILURE Node request 300-0021233738 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3a0bb541acfa4d18b7e7e557e05b26cc : NODE_FAILURE Node request 300-0021233739 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9462e90e45ac49a9a81ddff88b297197 : NODE_FAILURE Node request 300-0021233740 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"34c1ff172d11b931a971a9cb21d4b3409ed5066e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 05:17:02.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"11f7703eed08e72493e5410536065b44276862db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 05:17:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"278c2c5ef601384ab7fa6f1b33f7d15dab6d528c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 06:37:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/227f81b008954af68a21ac34c58236e7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25328c4344564e02a7969d307de9fad6 : SUCCESS in 16m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ee8c6227b1444c66bf817bd355f21d23 : SUCCESS in 1h 08m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9140323e0f5e441c9749218b302079b5 : SUCCESS in 53m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dca760901c14f05a1548a42f787a2e4 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92cf6742f6fe41d6811879a623c6fefa : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1d9ccdf71014537bf2b288f10fb747b : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c88517eaa28485c81755bbb0b6b682c : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04dec8cee4c2461d8efcccf44c6295ab : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3d5e4a441fc24737b5f5dc856294217f : SUCCESS in 23m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/292a4abb3def431bbcc728dbba537312 : SUCCESS in 1h 14m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/96b2066b131041aaa9fb6e9229f5b9c7 : SUCCESS in 1h 17m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/835e947dc188467485034fdc63512047 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0f9eb84f16c349d098f7fee2a1f44094 : SUCCESS in 50m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f650919088cf437db42c7d1bc28ef61b : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":7},{"id":"ae2788a1d3fce8e933458f27bf7e16a559b9e9ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 06:37:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"ce2315f9c6480bfa1fac4cd9e7b1b7316557e71e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 06:38:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f01e0a56fc3844089d07dae25412d515\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e673fec730d7400c8b413d9bed036d86 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7aa6450491644068300575062993c53b9413568d","revisions":{"f9b3e645ab11a7e9505dcd8b6c4b79ad6c185beb":{"kind":"REWORK","_number":1,"created":"2023-02-23 12:09:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/1"}}},"commit":{"parents":[{"commit":"b953d37dea73a09460936acef0de49339ba4aa09","subject":"toolbox: Remove dogpile.cache and pbr from u-c","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b953d37dea73a09460936acef0de49339ba4aa09"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 12:02:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 12:02:22.000000000","tz":60},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b3e645ab11a7e9505dcd8b6c4b79ad6c185beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b3e645ab11a7e9505dcd8b6c4b79ad6c185beb"}]},"branch":"refs/heads/master"},"a21099c6fe38316b28b9baf6567dc4f1a9b84928":{"kind":"REWORK","_number":2,"created":"2023-02-23 12:29:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/2"}}},"commit":{"parents":[{"commit":"b953d37dea73a09460936acef0de49339ba4aa09","subject":"toolbox: Remove dogpile.cache and pbr from u-c","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b953d37dea73a09460936acef0de49339ba4aa09"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 12:02:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 12:29:40.000000000","tz":0},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a21099c6fe38316b28b9baf6567dc4f1a9b84928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a21099c6fe38316b28b9baf6567dc4f1a9b84928"}]},"branch":"refs/heads/master"},"2762e59ffc79022bdbe0aea4b1071f6fd8a67d53":{"kind":"REWORK","_number":3,"created":"2023-02-24 07:32:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/3"}}},"commit":{"parents":[{"commit":"a627713720e1393a5296d9918fec4d9796a9b16f","subject":"Merge \"toolbox: use ansible-collections-openstack \u003e2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a627713720e1393a5296d9918fec4d9796a9b16f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:32:15.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:32:30.000000000","tz":60},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2762e59ffc79022bdbe0aea4b1071f6fd8a67d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2762e59ffc79022bdbe0aea4b1071f6fd8a67d53"}]},"branch":"refs/heads/master"},"3de6e455259b140867defe3eb12395ad002c63f1":{"kind":"REWORK","_number":4,"created":"2023-02-24 08:03:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/4"}}},"commit":{"parents":[{"commit":"a627713720e1393a5296d9918fec4d9796a9b16f","subject":"Merge \"toolbox: use ansible-collections-openstack \u003e2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a627713720e1393a5296d9918fec4d9796a9b16f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:32:15.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 08:03:47.000000000","tz":0},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de6e455259b140867defe3eb12395ad002c63f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3de6e455259b140867defe3eb12395ad002c63f1"}]},"branch":"refs/heads/master"},"d8911895130b539ae6013f7d9a9777338cde6225":{"kind":"REWORK","_number":5,"created":"2023-05-17 13:58:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/5"}}},"commit":{"parents":[{"commit":"d965b1b3da9b8c2780a91781834a46f5886bd9c2","subject":"toolbox: pin ansible.posix to \u003c1.5.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d965b1b3da9b8c2780a91781834a46f5886bd9c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:32:15.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:58:49.000000000","tz":0},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8911895130b539ae6013f7d9a9777338cde6225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8911895130b539ae6013f7d9a9777338cde6225"}]},"branch":"refs/heads/master","description":"Rebase"},"83d33123223e752d74265f2c379a07337b0ce566":{"kind":"REWORK","_number":6,"created":"2023-05-17 13:59:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/6"}}},"commit":{"parents":[{"commit":"d965b1b3da9b8c2780a91781834a46f5886bd9c2","subject":"toolbox: pin ansible.posix to \u003c1.5.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d965b1b3da9b8c2780a91781834a46f5886bd9c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:32:15.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:59:53.000000000","tz":0},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d33123223e752d74265f2c379a07337b0ce566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d33123223e752d74265f2c379a07337b0ce566"}]},"branch":"refs/heads/master"},"7aa6450491644068300575062993c53b9413568d":{"kind":"REWORK","_number":7,"created":"2023-05-17 14:00:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/874916/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/874916/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/874916/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/874916/7"}}},"commit":{"parents":[{"commit":"d965b1b3da9b8c2780a91781834a46f5886bd9c2","subject":"toolbox: pin ansible.posix to \u003c1.5.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d965b1b3da9b8c2780a91781834a46f5886bd9c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-24 07:32:15.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 14:00:38.000000000","tz":0},"subject":"toolbox: use venv instead of virtualenv","message":"toolbox: use venv instead of virtualenv\n\nChange-Id: Id20394d06747b7f56e8745f129149c3f1c492ca7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aa6450491644068300575062993c53b9413568d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aa6450491644068300575062993c53b9413568d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
