)]}'
{"id":"openstack%2Fkolla~876460","triplet_id":"openstack%2Fkolla~master~Ia4ce65b0c250360ffa5d7221686656763a400276","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-06-23 16:51:20.000000000","reason":"Change was marked work in progress"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-23 16:51:20.000000000","reason":"Change was marked work in progress"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2023-06-23 16:51:20.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-21 12:46:58.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-03-31 06:00:07.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-06-21 13:07:55.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-31 06:00:07.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-31 06:00:07.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-30 11:40:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ia4ce65b0c250360ffa5d7221686656763a400276","subject":"Revert \"Pin setuptools\u003d67.2.*\"","status":"ABANDONED","created":"2023-03-06 13:30:08.000000000","updated":"2024-03-07 17:56:58.000000000","total_comment_count":13,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"revert_of":873894,"meta_rev_id":"aa2b5f8400f2faa5fed11ac95ed4255e0d3595d4","_number":876460,"virtual_id_number":876460,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-25 10:07:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-03-06 13:30:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-06 13:30:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-03-06 13:30:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-06 13:30:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-03-06 13:30:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 06:00:07.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"},{"updated":"2023-04-06 14:30:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-04-26 18:05:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-03 15:27:08.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"CC"},{"updated":"2023-06-21 12:42:31.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-06-21 13:07:55.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"402c8cb691535a1228d3ec45d5270f4096f73aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-06 13:30:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48f1929275c8ed8a132692464f4c836f309caa9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 14:46:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c93db74e084ed893be4aae5713e7ee\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e8c25f6811754b6d937c279cc1bc3806 : SUCCESS in 1h 14m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8bba0dc71a0e42ee8e4ac13a47eeb07a : SUCCESS in 1h 13m 26s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0ff9fd024f7944a68865215fab2bed84 : SUCCESS in 1h 05m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52170b96af5840829082320d3498f37a : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56ced3402da9469ca61909511d012699 : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/29499281647441b5a50738dcf26c0748 : FAILURE in 20m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f38f662e96ff40fdadf76b9e5bfdb7c5 : FAILURE in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec0760b99954480836d93ddeb9625acde9334915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 14:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/714e939f5ea141229ad4e036b2ab9dcb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6ae595ddbf2e4ab493bf9e1e6fa79eaa : SUCCESS in 19m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c268b4e36f2b4fd8b1f6844454bad479 : SUCCESS in 1h 17m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a4b07ed31d9e4b9e8279c0d4ce3c398d : SUCCESS in 18m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1ea3085b9371400793c5e5e1faa54420 : SUCCESS in 55m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/36486925c5fc4a64997d3ec6746d8715 : SUCCESS in 1h 07m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9ebb1cb26d43459a078f515619bacb : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed9d32fc233d45a9b5d925921a663dac : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26fe2453253541be88a7c82888595024 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13688c6c701f46e0a5de118dfd84999e : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de83c0e60c84443cb26a07066495ce20 : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/af594dfe73c744368a504074a0e23753 : FAILURE in 20m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/29fc40598f734c929f2f48c24082fdea : FAILURE in 23m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2fa2da3024a248e5bad20fe5856595b8 : FAILURE in 24m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2b2fb9e522e14c8897f1faa6514fe84a : SUCCESS in 18m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d249bea14cb846ef833b6aacabedfb7c : SUCCESS in 47m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8e8ebacc315e4f9aa89ccafde524bf09 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"afa2b445129a10bf0e5faa1c12d303771fcd9624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 11:08:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"44b13a41e3a24afade48f4c6c6f58c5da2fffdb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 12:37:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7032b5f8c8f7485fa4879b17167204b7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/79cf914f7002413fb419de7e777ddfc9 : SUCCESS in 20m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7bfa7c7352754fa48d3cebfa61329dd2 : SUCCESS in 1h 21m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ff9183f5b98c4916bb398c73c2d84868 : SUCCESS in 21m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f1a032dc678f4acb87e7f9db9087fad1 : SUCCESS in 1h 02m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8aeb92b810434276994909aaeca07bbe : SUCCESS in 56m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2259201d7848c4ab437772533b1e49 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb1ef9b814d64bfe88f09686fe471e53 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/260355b22e9240ba8f1f7ec120ae5402 : SUCCESS in 3m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f30a05d76e524aa3abe66fd0a37e7af5 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e3ea5c84a4489eadd74aca656f2feb : SUCCESS in 5m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d36447e68cb34b47963df767daddf2bd : FAILURE in 18m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b15ee3ac2f7c4db6a00042a38d5f4499 : FAILURE in 27m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a4c07b681c55409696d0484a0fe7941d : FAILURE in 17m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d75753d04931494db008649c74bea64b : SUCCESS in 18m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bde32ba2941c4dffb00a7e28abc1aac3 : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/96ac82a629c6474396b94d5a4e2dce90 : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b1eeb000a20e3fc2eb6c4ae09b089d8716b90005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 13:12:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf25ca52fca4824af6ba3a8e042be7c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1afdd9d791904bf4b36860c60ad751a7 : SUCCESS in 15m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b5434ef8378149108fce48dbf1c32e02 : SUCCESS in 13m 20s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a662d5c01de4459b123beb851600873 : SUCCESS in 2h 01m 05s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c8a212883e9469ead8f5b0fdd66a37f : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf86112b2830459eaeb2f542e42239d9 : SUCCESS in 4m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cd4bd727aa7542c382eeaeeb81f79f07 : FAILURE in 15m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7d93301d3b9c4f98880fb38e635487fc : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db0fa8dfdf40fd617a2e2f84962ce65db4748f6b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-30 11:40:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b31cd5c664eb7669427e7401bf15af0f3cb0cd5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:28:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33aa72a03cfd49ccafb6c6c0eabd948c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d444aebf1384472ba47af8218beb9283 : SUCCESS in 20m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/356866e6eb14444d85de0a34ce3ea0de : SUCCESS in 20m 10s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c7ec4a34c3dc4795943d57bade519647 : SUCCESS in 1h 14m 07s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee31be172db24dc38237cb91dca4a520 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d741806b13f54663b8bd2ccc469b4cd6 : SUCCESS in 4m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d5151ee0ee4a4df8a902ba5ec405c228 : FAILURE in 25m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2e9c1b27668c4657b31a007dfc6ac204 : SUCCESS in 1h 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0502052b878c58181ef7d57de7fb753711ef3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:31:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c208f4007cac4eb7a4ea7cdf534adf0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1edb53963a9645a79d58a934841efd7c : SUCCESS in 20m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2b02177812974b9e837a4be800133323 : SUCCESS in 1h 43m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b02c19e1a157452693631c83d451f9b9 : SUCCESS in 20m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/976a1661fdd14e6d85e3eb322adc4c9d : SUCCESS in 1h 13m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4a1a58583ec94e9bab289f1efbc25aec : SUCCESS in 1h 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe96b0823084bdabd1d10a9e7e56990 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baa9c46eef0449189f7a656c9d35b738 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01b0fca4147e48518ceb548de92d4cc4 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8563dfc511504625ae59c5dc92ed50d6 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f0b132b28a486fa0ac432aae11dfc3 : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6776d7efa4a74467aadb56c48a118adb : FAILURE in 21m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4f01f4aae9a549499e12e7ff401a2a51 : FAILURE in 24m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/437d9691081647e59f839a358c09f88f : FAILURE in 26m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1342b7b2642f470d981a17129eba95b2 : SUCCESS in 17m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bfe80550ee5f414b82c7378e073ad6ac : SUCCESS in 54m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73a84391f32640499fac235eeb5cbdbd : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0adf333ad54601603273ee026568297bca2e426b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-06 14:30:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdfe30a05a03e7434c1f0ad92e4d3d6bfb718166","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-19 16:15:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"745dd0a5ef9b18e1fb2969db87ae32b6b1d416b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:26:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25ca6373646740148e057b578ff1cc46\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3868802c60614f2ba50ede928649bfc6 : SUCCESS in 19m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/54bb92424abc465ab3f221adbb2b500b : SUCCESS in 57m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ea231ab167e0436a93fbe9dddf5a967b : SUCCESS in 17m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b2cff9eef7fa441a899252f152bb6bc4 : SUCCESS in 53m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a7404fda8d054764b94b156d342e773d : SUCCESS in 1h 01m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1f647f1f464086a8dd4d7c1eb95333 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cab6950390dc4e978ed1daa4f2ad8e1b : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4457862c5fd94a1889e840322485924f : SUCCESS in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f243cc07148e4e8981c81b98fdd5d10f : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097e7abefa644f71954737e8a30ec518 : SUCCESS in 5m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f3dc96b174d44682b904cf0ac845fc67 : FAILURE in 22m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9c4ba2589f324734b277f92815cccb61 : FAILURE in 23m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5a63bd6b8e024d8c935bbab091843839 : FAILURE in 24m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cdf7507a3c14416ab429dd77cc9d7a1f : SUCCESS in 20m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a2bc29f0c1324dac9d3be159f493307d : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c6f45d8d53b34b6da2aef486727d3abf : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6a476cab4731d92be6c7f1c395d1568a266468d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:33:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1080b6f13d4a6da74680d682741e1e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7a0b4b1e22f149e783bdac89166ae79d : SUCCESS in 17m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1bbb54325aab4e799922257df2556873 : SUCCESS in 18m 26s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2795353db4524434bffc8697779678b3 : SUCCESS in 1h 04m 33s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff0533762e7e46159fbc0b6963043e2d : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c288d2e29fb414dbd388f8834539ae0 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5be6f4c81da74e7b80a11889f5c13557 : FAILURE in 19m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/debf24e70762431397cf05c19bb1d746 : SUCCESS in 1h 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b97b1b68fcce9825ee5a210e98a4fa350a3b8f4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-04-26 17:54:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"701fd5ca07fe636dc6358ac4d18d3491ac7ff8ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 19:15:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7d314c827fe485fb0f19bbb39541cb1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/17588338000c436497abed5bf2bd1d19 : SUCCESS in 19m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d247b666eeb947e88c25e94738e01cd2 : SUCCESS in 1h 10m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3fcc757b03884273aba284f477ee498d : SUCCESS in 17m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bbfb77fcee5d4e5a9a052c0e83614d3e : SUCCESS in 1h 06m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5aea4d3fb51445218091e6232c0683ed : SUCCESS in 1h 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442cf49e657b439ea2efffba72bc02fc : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88d5f48a6e2c41e185ce92a0e731c3d7 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8168f75ef57741ee8d2497f2974fedb6 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b0517a8dad2430d8f2794ee68c1a83c : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b0f9e5acde846fa9b6090ca72ba672a : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d23948b8f1d0462a8c2a2f8c1e449808 : FAILURE in 17m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2f714baa273849ea996fc3ebbec563d1 : FAILURE in 21m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c9be7bddabb74fa5b3f411ad4fc21ec4 : FAILURE in 20m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/288f0ac410aa4b26b35b003c01febaec : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d4f88f2a3c3a4308bacc8782616625a0 : SUCCESS in 58m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ce0774bc75c74f8682486f91a7661430 : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3f8817b703b1a37e6b468af767b416a7b599634d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 20:04:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfec96504da4843951eac0a2ab86c18\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/279be5c47bc544d99004623ce76c041e : SUCCESS in 18m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/19415a1c5d7140aaa476135ac7081b69 : SUCCESS in 1h 04m 36s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1029717ade54482d8bb9d4644e2335ed : FAILURE in 2h 08m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3269d6a66bc42bc912e3f7d522aab7a : SUCCESS in 2m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e97f737a900c4d5e97de00dc98716a1b : SUCCESS in 4m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9d920b16fae24fabbe09a1726f162032 : FAILURE in 1h 06m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68f34c20d8cc4d288c9098e91ec451f7 : SUCCESS in 1h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38fce45a48da8aa4605c5d4e284597c51fbddf0c","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-05-03 15:27:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46dbc7b001990c3a9554050b408f81cef2a27b2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 12:24:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fbb6395e58a3f440cc6eb58a25efb979ec60a8df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:53:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffac7c56528c416bba73ac7e8e526b6b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/21accef68f1049c2899ea9c36deacbce : SUCCESS in 1h 25m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/220aee5d31e94be9b2fc34de20ab813c : SUCCESS in 20m 59s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3477593a790f4b6cbcd62c3f9cd9ae17 : FAILURE in 1h 13m 17s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fa1af0393a541d4b2e11e437c4ecf98 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56691701e26f453b9922397a79450c8a : SUCCESS in 3m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e2ca4c2732ae451ea5d15dae1134e647 : FAILURE in 24m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6db32687282e4c418e0bbf0ef64762e3 : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d43fa4be1c0aac7c517eb3d4bb57b66011f4dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:55:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a85badfc7a454acebf68f5d690e2c7b3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/09eccdc06d8d4af0b0832071cda02e98 : SUCCESS in 34m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e644296746ef423c8d4ee3632111557c : SUCCESS in 1h 29m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f4fccce11a664b2bad8d2f77f3ec86f2 : SUCCESS in 18m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/359afde4731649c6bef172236d7d4802 : SUCCESS in 57m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/09859dbd8dd44838911c85997438fc84 : SUCCESS in 1h 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186a4833b4494cb186b1e2763a7ef290 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b289e1733c004c8a824784354deebee8 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3171bfdff8c34c1ab6ab6263528c5fd9 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00cb384cbde445bca49db6a01456c97c : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25635b02d79f43738243fb34824624f2 : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6e779b491ea74b74bae77a7040cca869 : FAILURE in 25m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9b85bb5142c474d88c80d9a2edc00d3 : FAILURE in 22m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bd9298c612bc4b2fb653542ce72ffad0 : FAILURE in 24m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cc605223c35c4b8e9155dcafd96ebcc4 : SUCCESS in 19m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1288c14bc9fd405dab0d9410af759305 : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bf73bb7f9c42492a956e97b8fec439be : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":3},{"id":"ade12643aff43f4fd3db26c7d48fe623e5d3d728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-17 10:32:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"066e7cce44a810db183e83b825699ce537a02212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 11:50:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"81921b5e3e129c555c0b86e9cf6faa63ec855ac9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 11:51:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c2ef234d2a6255297351a493c7bf15495ccf3c22","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-17 12:11:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"123a2fdc778606604d8aff98d09c4d9cd0c2e990","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 12:16:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"568c90471848cf70ba448387d40ea9b38df50b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 13:10:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5fbe1de44cf4fb397d505d6ada6a56f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/882d0f44c2ff4476bf70bf1549f04e60 : SUCCESS in 20m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9f74068a25944db89326d7c10dcb6e38 : SUCCESS in 1h 00m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c44fb813791442afa04a77d7ef96d244 : SUCCESS in 16m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1e3b212687b644cba47f100a3b450c73 : SUCCESS in 52m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d5b21d692a754114bdf1c69195c2947b : FAILURE in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127e5b85d40e47c1aff5ea79bf6c8016 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85700be216f4689b4e6f94e095e6e91 : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1c9da84ed5a4db6bc7deba0a9427577 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/778cbf09d1fc4d3cb525ce43c117e204 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31da476f45fe4f64bdfa68692d08aadd : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3d500eeb4b458da25775ca89c1210e : SUCCESS in 5m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/730cdebb1a864d6289d0afca99d6e39e : FAILURE in 17m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/684269dfc7de4373b5cf64517b556563 : FAILURE in 32m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/94fe2068e777414fbd79442fc2b97cb4 : FAILURE in 23m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f371aee44f1e4596b72345fb7a3bb15b : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b3aa885d914c4a5aa3c0e47ba65f4171 : SUCCESS in 57m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/34b10bad0df34000a32298ea27fab9b6 : FAILURE in 58m 46s","accounts_in_message":[],"_revision_number":5},{"id":"d4e75a7210112f429dbcf4f6a6ae6ed10e61d9c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 14:37:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ca81c4f4bd4ad1bea59563c5ae0d96\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/511c9ce6d7a940bf916cc0bb3af6c6f4 : SUCCESS in 1h 33m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66e31d2c4c2341d2bca5dc6c014fc14b : SUCCESS in 22m 34s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/856c3c6c92a84db5a9ce52c0d0b79669 : SUCCESS in 2h 44m 38s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a63db639adbc4a1bb845a106ac14ff8f : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b049b30c9e34ab3b0d52ff4695fff60 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aea58edd7c05414c81279e0f0511781a : SUCCESS in 2m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/922724f4e8934ec78469f4471c761fd3 : FAILURE in 1h 32m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/13a0c480304745dfb234631580dc6c5f : SUCCESS in 19m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"09bac0d126ddd4cb890bd378bab1e9ccb707dbf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 07:23:11.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4ffc418c16e1e1a3227fa064bee8f47a3b60e4e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 07:54:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ff5010f8244a53be9a3dec4e84ffb5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2c6a48072b9f41a3aaafd8382da73c5c : FAILURE in 30m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a6ebabb51874961bf3669569da1a28a : FAILURE in 26m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a6588ec38e5b4b3083d81b1a743779df : FAILURE in 23m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f54a1b7ae50b499da62697ff63892f3d : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7774368fed543ad911c278cc07bb538 : NODE_FAILURE Node request 300-0021492542 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cff9e6446ff1467994d7fd230eac975f : SUCCESS in 3m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/88fbe5fe486340deb79d9292ceb347bf : FAILURE in 24m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1380e2716db14bff8c1cfef331ef954e : NODE_FAILURE Node request 300-0021492545 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0050810a98c1cc4a9f974652abf179709dbbb0ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 09:22:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/483bbe527a794beab13f411b9cc96945\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ea06eb32d39c423cac1d399d64fc8bf5 : FAILURE in 37m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0672c65ad6da49a08e51be7f57e39dc7 : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d4a7f952cb2415087b7cecba73f1f1c : FAILURE in 19m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15362a2d6dc44cb299b9de614125bc10 : SUCCESS in 54m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cf306e51e5de4d889ef4e0f825817ba9 : FAILURE in 50m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8867af5a0ef24f42a7187e2249769a9e : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41d9eb1878cf40288a8e3b29aac70bfa : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c370a8e4d64432bb029014ad60023dd : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b75701f3c1f42c0803295966b895609 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fe7b55fa83349d79f6855750061a07f : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f824126a210b4423a0884b5affcf1dc1 : SUCCESS in 3m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/615a642a5dc24587b702024261a061de : FAILURE in 26m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/246b12a2da90448d862f0bcc5475813b : FAILURE in 23m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/86667a7ad63e49f996002481cb0270d0 : FAILURE in 21m 18s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/eb4b1e1b9d3a415b981588038ec8760e : FAILURE in 22m 37s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/750886e3233848ca8fdb9659e928af44 : FAILURE in 24m 49s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/10768c0812ed47f3b884bdf25b042277 : FAILURE in 26m 44s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1bf01114bd38409286ad15003cd23fd6 : FAILURE in 24m 40s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8d7e56a2e31c4ce99d6efde20b6187de : FAILURE in 23m 35s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8d302e284d4b4403921253709d28dde3 : FAILURE in 25m 18s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/968cf89f44de452d9b4f964b67fe44fd : FAILURE in 25m 11s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7163bc57ef4e428eb906c28f38dadb29 : FAILURE in 25m 16s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e86d3a1441404e6fa6741587a8cdc712 : FAILURE in 16m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fdf2b48fcf8e43b083ce26fc9d470191 : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bc25bb44c60c446abb90689f673f2fc3 : FAILURE in 49m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0cec06c4e47242798a9c34e8650c723a : SUCCESS in 1h 24m 55s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5031993cf99d4651ad5938cdf20f8e7e : FAILURE in 1h 04m 12s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/024c799176394adb92558c04a989c539 : SUCCESS in 1h 27m 48s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/dd20d421297a40e191b2985c983620f4 : SUCCESS in 1h 05m 57s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8e6b0b5e6fc5474da0c3329e33817e09 : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e909225f12e144329d3dd166135ceb1c : SUCCESS in 1h 01m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/eb9e520910e447c78134dfe22af9696d : SUCCESS in 1h 06m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a3c985aabb5e427098411eb67398a8d5 : SUCCESS in 1h 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ee930a49f3add846009a440cf5714aadeaeaf3a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 12:36:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"e163b074ac3c679e6d1d8ee4aa2c060d717dd78b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 12:36:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"03ceb2abd58be9beb9eeabce9457ab2fe852ff7e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-21 12:42:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"34ed9dda2abfb26a1d7053d1217879c1b460675c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 12:43:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ac9f3a6dc896d2344f02a911e4dd92c4a1b1b330","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 12:43:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a9a1d2674eb1c365143658f71664f0bb4549b62a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 12:46:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fd2124f7892a2ea9e21b9d304c4e65b3cea8c44d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-21 12:47:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"8bb8666290b602484799c4a93bd306f81b0f3ad7","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-06-21 13:07:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"89f844ff1bed04bc40582e1bad5cf6bee59edabb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:15:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139274e6ecef4d01b283435aef84d652\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/85deccb590a042bfbaf0184d0a3b1e7e : FAILURE in 21m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/87be5a06c4884111871e97f3d5b7096e : FAILURE in 23m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62e7f8570a67428785a8a5f866f30ed4 : FAILURE in 36m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cca99d9acac7496e83e1c5396ecca61a : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8cad843879e45e3920261939e3fef62 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d253973387e423aad09599d4dceb31d : SUCCESS in 2m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/89bcaa924fa74e899cb4e30d5789f24b : FAILURE in 19m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/636f0f298ea04c8ebce90cd879b46bc6 : FAILURE in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c0629b0d45fe10cbfaa63b4d35f1866312c63e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:52:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22670705b9894708a9c38acfe6535dfc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/165bb99c81b84889bfd89dc8e16f63a8 : FAILURE in 28m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4258abda069b482bafe6897c89715f82 : SUCCESS in 1h 00m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77270b0030bd4f2e95c329755410db36 : FAILURE in 20m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f4e5e6ff535f45c8980411e8d9ca4ba4 : FAILURE in 20m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/491ebc42fb644a71918d36ccf83231e4 : FAILURE in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf2076d0f88486aa6ad1f20e7d4896e : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53fe883f7513423787ce6353cc711044 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be17312601d942f9a9c4284a653ab864 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7fef36971e44de49cf4d20bcf216d6d : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a54e84a314b5480e88f81fab36ec6dfe : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a08c4c99b5d42bf9bb3570e66005917 : SUCCESS in 4m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/50626d473bf04c29a9d7f8f5adf68201 : FAILURE in 26m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c88c831a15ff429ba1c305cb489121b6 : FAILURE in 23m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/326b153b7250442cae15d7e107592594 : FAILURE in 20m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8d51c866a3f34594a5cc132645a59043 : FAILURE in 23m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ab6664bb93f84d97990b4a6f44be29db : FAILURE in 21m 49s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/5a49f10f1f86437ba36688856abd7402 : FAILURE in 24m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ef6337bde19b4de09984b71f7600e88d : FAILURE in 25m 12s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/fae625653d384f36b207a06be8c0279e : FAILURE in 26m 58s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/de79aece2c9644d3b9ae7027fba01d63 : FAILURE in 24m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9372caed42bd49a0801b77246cb9bc27 : FAILURE in 27m 20s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/14f3c33e98644cffa385fd268e38453b : FAILURE in 28m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f78013cb452c4ae3bdba54c6d0a86635 : FAILURE in 24m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1e559c40ee334b78b87515d86ded789f : FAILURE in 17m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/99e10bb2ee414c2bbd81df557be3fa8e : FAILURE in 15m 21s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/72b0a6ede437402e9401781430d3a3ee : FAILURE in 19m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/76a7e776793f472ca74316d8c697a16a : FAILURE in 18m 04s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/9d83c217521a4aa7a605871a794da41f : FAILURE in 19m 15s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a9d0dd670ea64c0d9310512bc4e3e715 : FAILURE in 21m 08s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/65c0de4a05a641cda244e4960cc00b94 : FAILURE in 22m 29s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bfb26ee372c6476f9c9db9d55dba88ec : FAILURE in 18m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1b0e169e374a4bc293290a86051f18d7 : SUCCESS in 1h 12m 22s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1a5f5794a80942e9bc785bf74adc476b : FAILURE in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"541d2edf35a0c2177705db8c62fc1bc360e8cb9d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 16:51:20.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nThe following files contain Git conflicts:\n* kolla/common/config.py\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"441e27fdc69ac2c510717792a3e93b26d16b838b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-23 16:51:45.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"eae9e2ef8785793cf4281df35dace3ebc3e767f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:12:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c1bf60dc978420485a997058fbd8877\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/87aac041c8384bc88eea9b45bc5b10a1 : NODE_FAILURE Node request 300-0021529121 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3186bc5f8ec64c98a475543bde3ec9ed : SUCCESS in 17m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f285d07ab1d142a3813038c2dcc728c1 : NODE_FAILURE Node request 300-0021529123 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66757534b1e34775922d8339de0a485f : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3df1ce4e4da4c60b792b0c01a6cb836 : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1601d39f80e04c5da300a7e28aa99471 : SUCCESS in 3m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d5c4d2d3022c4c8e9968f7cb52a93c16 : NODE_FAILURE Node request 300-0021529127 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/051d85e8b015410dbd0feed15217c64f : FAILURE in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6f5fc20fc4f1f0e44fb6535974426225ae3571cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:28:27.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12d420f17e444aa59c86cac0a15127f1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/275f588f6784418898579a0929d660ab : SUCCESS in 37m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff57068c9a9c4547861c50faa6325023 : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cd34c7ed9f014c4e83d38962cea01888 : FAILURE in 35m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6368ef66572c489bb9144ead729e2bb8 : SUCCESS in 1h 06m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e78159d2adee42a8900699372c596304 : FAILURE in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397a5f139086441e8a68af4ee3bd259c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b7e3335dcdd4f449a907fcea7637daa : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea481cf0dd92498e9774cea61e210c98 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84f90da85c514db6ad70a23f65be9abb : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/470ff3bd94364c82873b7c96d098e92d : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc59889fa42d4ba58b7e0c099b682f66 : SUCCESS in 5m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3d54cd4571d34e46a5512bea0b4c1bce : FAILURE in 50m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8daef5d3a0cd48b3a6c22a2aefe52f0b : FAILURE in 36m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/484be8dfd7b240a3af6de1b626a2ce90 : FAILURE in 50m 04s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7fc79549d31f4c69b1e76dfb9c8b5a83 : FAILURE in 48m 53s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/29b4ff3c5ca04e239ab1526492f1ff29 : FAILURE in 48m 55s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/9c1399c28d8a4d7da253bcbc78cf1b6d : FAILURE in 29m 41s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9ca7027053d642f0a917c43bc462f246 : FAILURE in 33m 33s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/1c1ac7aacb8341688cbac0cb0bedb3e1 : FAILURE in 45m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bddd4ab73a254cfea0e98db0049401ca : FAILURE in 50m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/10fc53ae780f447d8c5107e685d03bdd : FAILURE in 33m 59s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8f36d4c341454c01a00247202bbec225 : FAILURE in 51m 17s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9d7dd0dcb16341d69149bb9f7bfc076e : SUCCESS in 23m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b0ffedfbdc024d9f90c6a6ed6daa3065 : SUCCESS in 1h 24m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1950fc89a6f94f918ef6694b2e1f1f78 : TIMED_OUT in 2h 32m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4868dd1bde6f4ee7902f5085c6e42471 : SUCCESS in 1h 59m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/00bdbc5dd5cc4f11a1c3e51c9989d42e : SUCCESS in 2h 05m 48s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/04748455804e406b80fbdc2c61f72bf8 : SUCCESS in 2h 00m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5fa08c54264946a7a978318323fc9713 : SUCCESS in 1h 35m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b2cb2537b3584f03aca57748e2592634 : SUCCESS in 1h 15m 04s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9e73cc72d6b14dd1a766e81fd726ac11 : SUCCESS in 1h 25m 57s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2106c137cb324d0dbe22c9c2a13ea23f : SUCCESS in 1h 37m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a9e1f615f2694548b53b2925f0ec80ef : SUCCESS in 1h 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2c500d85ca564976282662fc30cbadda9f596fcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-25 08:18:34.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nThe following files contain Git conflicts:\n* kolla/common/config.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"5383a8816fb9cc6484a0db353e0721ba3cf85bae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-25 08:18:54.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"4f3bd7057b229b7f891673cd1a7bcf761205f768","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-25 08:19:07.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"842689597d23de612e9fe251f05abfb5de3d8519","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 10:04:11.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d711a5e5f5fd4738837736b5b5944756\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f5ba3f02521e42e49041b5989d0067f5 : SUCCESS in 1h 41m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1792589c79d2496e9897ea66225bd720 : FAILURE in 1h 37m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ff7bd7237ad4e04b7068541452ed86d : FAILURE in 1h 17m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c54e27108d847a18732ad85d24e310f : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e593dcdf212b43339cf7371f8ba353fc : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/90ed5afe29b346d493fff4ccb022c8df : SUCCESS in 9m 59s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/904f9c561681487a96689c5345db9088 : FAILURE in 1h 30m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/334b5d082d6e490a8bc1ba583653be26 : SUCCESS in 1h 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d5647b57bc8b2634049d420ade3b0a9afd072996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 10:07:36.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a6586dc6e474b549927b14bcbe993ab\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/22bc641824bc48d0b15c78b31b0d3e64 : SUCCESS in 24m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0deba36b979a4d51afa33b2ac7c8c366 : SUCCESS in 1h 16m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/902dd2e0dbbb4c1f9cb17f34e1fd04bc : FAILURE in 18m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d716bae01f4a4ffd8910eead7003f897 : FAILURE in 34m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5260e9e0513c4e3a9734a806394fc374 : FAILURE in 19m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f87187140b474abf99101ef093812dde : FAILURE in 33m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c3e7f6b3df45f697f9fb2272421699 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e587c2d9da904d10bb36dece3584c6f6 : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e472884d44be473eb7d2c2035c59fbd0 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75a4dcd0a5114ea0b6740d7c64fb8ba4 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72047333b5e94551b92cbf966612b676 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/14c1a6b130ec413a9ac8fe662fd39bbc : FAILURE in 23m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/052c42bd343149bd93b13228a3cdca69 : FAILURE in 1h 10m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d86be21b84db45bab18138c9c1ac1fbc : FAILURE in 21m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8253d307f34b446db585817bb0b94000 : FAILURE in 22m 19s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a0ff6ad475c64b3aa78b254de7b43bd0 : FAILURE in 25m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/832e67d3eb134f00b11a2fe7e3afd705 : FAILURE in 22m 48s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/54393bd59eea427089c2350ace46b57a : FAILURE in 21m 42s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/dba6d5eea0204121b93609d049278a65 : FAILURE in 23m 05s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7784db3ef5f54d8398f9742c56dae52f : FAILURE in 22m 13s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/dce3668d21d54a7aaa224a303bc510cf : FAILURE in 22m 35s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b0a155230e5345ca9052886c110d93b2 : FAILURE in 23m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/086dadd35e9b44d79ddb09e762264a0d : SUCCESS in 16m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/75cbe66892d44406817968f56417d25a : SUCCESS in 26m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/163cb0decdee4e1fa939ae15b38dcfbc : SUCCESS in 1h 00m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/99b61c16a2f74c128b782ade12346f24 : SUCCESS in 1h 05m 46s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3f1e95804daa4f45b3bdb74d204b4ddb : SUCCESS in 1h 37m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7a8534b7513744619cc8944996b0fec9 : FAILURE in 1h 09m 11s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/092a29b7f6d1436ba206609af7327313 : SUCCESS in 1h 17m 55s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0bde96f9df0241dfa3f92718715f43fc : FAILURE in 1h 05m 49s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f3e9755a04e14ab4856ac37b88615be3 : FAILURE in 41m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6c9444ceb64b4412b17b0afa2c9058e1 : SUCCESS in 1h 11m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8eb17379440a4f30ac9f70ea0c2f793e : SUCCESS in 1h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"aa2b5f8400f2faa5fed11ac95ed4255e0d3595d4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 17:56:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"33009c971d1a5a00891f3c156eaa05152d49808b","revisions":{"bb6a0dedf4f6a6ed98c84ee5cd9dc2e0b8c34031":{"kind":"REWORK","_number":1,"created":"2023-03-06 13:30:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/1"}}},"commit":{"parents":[{"commit":"ca0238598a37c4a174c7b8d55b64c15bb34ef58a","subject":"Pin setuptools\u003d67.2.*","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca0238598a37c4a174c7b8d55b64c15bb34ef58a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb6a0dedf4f6a6ed98c84ee5cd9dc2e0b8c34031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb6a0dedf4f6a6ed98c84ee5cd9dc2e0b8c34031"}]},"branch":"refs/heads/master"},"82483c62d167702918cbd00587093a6152815f51":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-21 11:08:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/2"}}},"commit":{"parents":[{"commit":"32170b008547099a90473b6ecd1e767eb636834e","subject":"Merge \"Add Dockerfile for neutron-ovn-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32170b008547099a90473b6ecd1e767eb636834e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-21 11:08:24.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82483c62d167702918cbd00587093a6152815f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82483c62d167702918cbd00587093a6152815f51"}]},"branch":"refs/heads/master","description":"Rebase"},"560de4cd4ae0dd8f4ffe7bc8903f519d2eaac0f4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-04 12:24:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/3"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 12:24:45.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/560de4cd4ae0dd8f4ffe7bc8903f519d2eaac0f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/560de4cd4ae0dd8f4ffe7bc8903f519d2eaac0f4"}]},"branch":"refs/heads/master","description":"Rebase"},"0939299c77288192f034101c9163ab682d98b459":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-17 10:32:31.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/60/876460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/4"}}},"commit":{"parents":[{"commit":"d965b1b3da9b8c2780a91781834a46f5886bd9c2","subject":"toolbox: pin ansible.posix to \u003c1.5.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d965b1b3da9b8c2780a91781834a46f5886bd9c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-05-17 10:32:31.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0939299c77288192f034101c9163ab682d98b459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0939299c77288192f034101c9163ab682d98b459"}]},"branch":"refs/heads/master","description":"Rebase"},"6b306331a47762b88eabe7a5078c3be134f1146b":{"kind":"REWORK","_number":5,"created":"2023-05-17 11:50:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/5"}}},"commit":{"parents":[{"commit":"d965b1b3da9b8c2780a91781834a46f5886bd9c2","subject":"toolbox: pin ansible.posix to \u003c1.5.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d965b1b3da9b8c2780a91781834a46f5886bd9c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 11:50:31.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b306331a47762b88eabe7a5078c3be134f1146b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b306331a47762b88eabe7a5078c3be134f1146b"}]},"branch":"refs/heads/master"},"0259f8d0f39bf72e0257175e384a2dc7d048b408":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-06-20 07:23:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/6"}}},"commit":{"parents":[{"commit":"232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba","subject":"Merge \"Fix cinder-volume: Add forgotten lsscsi rpm package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/232e5c9b64aa3e345abd96aa5b4c4f9b3dda1eba"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 07:23:11.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0259f8d0f39bf72e0257175e384a2dc7d048b408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0259f8d0f39bf72e0257175e384a2dc7d048b408"}]},"branch":"refs/heads/master","description":"Rebase"},"874b647daa2a76354a08701e73fa0e24566b0fb5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-06-21 12:36:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/7"}}},"commit":{"parents":[{"commit":"60603d240cf26fa9bed8f7883ccfd5ae51c36f2e","subject":"Bump proxysql version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60603d240cf26fa9bed8f7883ccfd5ae51c36f2e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-21 12:36:45.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/874b647daa2a76354a08701e73fa0e24566b0fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/874b647daa2a76354a08701e73fa0e24566b0fb5"}]},"branch":"refs/heads/master","description":"Rebase"},"27852cd701f1a003d80ae7826de1bc7caacf2f05":{"kind":"REWORK","_number":8,"created":"2023-06-23 16:51:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/8"}}},"commit":{"parents":[{"commit":"b33167972d6e17b2eaab85e72043e3b4efa4b789","subject":"opensearch: drop unnecessary workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33167972d6e17b2eaab85e72043e3b4efa4b789"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-23 16:51:20.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27852cd701f1a003d80ae7826de1bc7caacf2f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27852cd701f1a003d80ae7826de1bc7caacf2f05"}]},"branch":"refs/heads/master","description":"Rebase"},"812f6c1cd6af39731ecacf02d0aa5a249e8e7300":{"kind":"REWORK","_number":9,"created":"2023-06-23 16:51:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/9"}}},"commit":{"parents":[{"commit":"b33167972d6e17b2eaab85e72043e3b4efa4b789","subject":"opensearch: drop unnecessary workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33167972d6e17b2eaab85e72043e3b4efa4b789"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-23 16:51:45.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/812f6c1cd6af39731ecacf02d0aa5a249e8e7300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/812f6c1cd6af39731ecacf02d0aa5a249e8e7300"}]},"branch":"refs/heads/master"},"8b9a2cc757c5f7e80b70a71479d028b778e99e34":{"kind":"REWORK","_number":10,"created":"2023-10-25 08:18:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/10"}}},"commit":{"parents":[{"commit":"e77002146990c6df2c961e07e1da41440fdab122","subject":"Merge \"Revert \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e77002146990c6df2c961e07e1da41440fdab122"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-25 08:18:34.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b9a2cc757c5f7e80b70a71479d028b778e99e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b9a2cc757c5f7e80b70a71479d028b778e99e34"}]},"branch":"refs/heads/master","description":"Rebase"},"8d7d30798af8819b7a018aa42fdb2bd6ff9ee2b0":{"kind":"REWORK","_number":11,"created":"2023-10-25 08:18:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/11"}}},"commit":{"parents":[{"commit":"e77002146990c6df2c961e07e1da41440fdab122","subject":"Merge \"Revert \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e77002146990c6df2c961e07e1da41440fdab122"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-25 08:18:54.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d7d30798af8819b7a018aa42fdb2bd6ff9ee2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d7d30798af8819b7a018aa42fdb2bd6ff9ee2b0"}]},"branch":"refs/heads/master"},"33009c971d1a5a00891f3c156eaa05152d49808b":{"kind":"REWORK","_number":12,"created":"2023-10-25 08:19:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/876460/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/876460/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/876460/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/876460/12"}}},"commit":{"parents":[{"commit":"e77002146990c6df2c961e07e1da41440fdab122","subject":"Merge \"Revert \"CentOS/Rocky: use CentOS Cloud SIG repo instead of Delorean\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e77002146990c6df2c961e07e1da41440fdab122"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-06 13:30:08.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-25 08:19:07.000000000","tz":0},"subject":"Revert \"Pin setuptools\u003d67.2.*\"","message":"Revert \"Pin setuptools\u003d67.2.*\"\n\nThis reverts commit ca0238598a37c4a174c7b8d55b64c15bb34ef58a.\n\nReason for revert: 67.5 is out and pkg_resources should work (but deprecated).\n\nChange-Id: Ia4ce65b0c250360ffa5d7221686656763a400276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33009c971d1a5a00891f3c156eaa05152d49808b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33009c971d1a5a00891f3c156eaa05152d49808b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
