)]}'
{"id":"openstack%2Fkolla~877757","triplet_id":"openstack%2Fkolla~stable%2Fzed~I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-20 10:41:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33","subject":"Test for unsafe files in tarfile.extractall","status":"MERGED","created":"2023-03-20 05:21:09.000000000","updated":"2023-03-20 10:42:27.000000000","submitted":"2023-03-20 10:41:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877757","cherry_pick_of_change":877611,"cherry_pick_of_patch_set":8,"meta_rev_id":"12413d12462d176415a25fe2cc9244243db131a8","_number":877757,"virtual_id_number":877757,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 10:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-20 06:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-20 09:14:30.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-03-20 09:14:30.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 06:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 09:14:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"323175bf599aa72035a5e1e82a2b909f85db6d27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 05:21:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6d5db5fb28e5d05128f565be8824748908db4bd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 05:21:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d45a1d371deb08ca66285ae6d40981db5783b1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 06:38:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc2ff5ca90ae4658b0b30c804f42d4b9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/154825dc8b3040f0baafcb3cbe23f620 : SUCCESS in 21m 18s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/8b407bb568ed4803825eb495861e0a12 : SUCCESS in 1h 05m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8087ca5bc857409689e2efba4f599d55 : SUCCESS in 20m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/40c3c0ba32fc45119a83bd73a3ac8ec5 : SUCCESS in 1h 09m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1743e8edd26d43ba93e97ba8ab8c3480 : SUCCESS in 1h 03m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea6358d5b314ab4b99c585f8f82d406 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73976b2282bc4be6bb170eaaf146cc2b : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eb029665584445690c2a896ea5f0234 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df86daff75b04602a25dfbc0c9b36c99 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c60aff8a4e47b281b38a73649b614a : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5d6cad3ff82248dc8afac7351764b469 : SUCCESS in 16m 32s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bd45f86f25f3464b88747cf613819624 : SUCCESS in 1h 09m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d742ad6c7e9d47ce9adb785a8925c92a : SUCCESS in 25m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dde03f715b4d4f8facee8d81f9fe2f20 : SUCCESS in 52m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc646fb5a9674ad296717982195a88f7 : SUCCESS in 1h 10m 33s","accounts_in_message":[],"_revision_number":2},{"id":"c16483e90f4a6b54ff123f4535b8af2f18beacc7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 06:49:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"84856eb90570de514d248ff24311057096bb82d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 07:35:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6758ae32cf684633be1b80b04028b3b7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e2cd78aada75445f8497bf9ff70990e5 : SUCCESS in 2h 06m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4e7d552221614967bc847f94a03e12d0 : FAILURE in 19m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08655db2e5f4440c926f59af7e876b3c : FAILURE in 1h 00m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbb03e4fb5be4ac09d4c3554f42f8e9f : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/443985aa3149459abd81b4a1f7f899d4 : SUCCESS in 2m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c8cf3bbda54644e2a3599c955e9c002f : FAILURE in 2h 11m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/baf1e77981a64f6c91b612bcb72fae69 : FAILURE in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc241e30b812ee194970324a1c1de09af50c5070","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-20 09:14:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0cdfa05b80c2ec3d3ee6611b27cec89bb709845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:14:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c367b441fa58fe1726621dec020758cdc278edb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:41:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b70af840fb34e158b646531d0842634\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d3b481b257d741a485c430f2b5bc90a2 : SUCCESS in 21m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4f70e935ec084486bd06ece328264423 : SUCCESS in 52m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6eba8c640fe94a149f8470e7fe605ed1 : SUCCESS in 1h 21m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a65246b17b46a6bbae5837a46d87b3 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6847c797ad994044b86fa5b706dff872 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c43308e088f842d2b5eb50f60c2f6738 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e07932e0f5c4de3b1b7662aedac3845 : SUCCESS in 6m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bab18667fbc54f2391d0bf19bc645941 : SUCCESS in 24m 46s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ae61d16a81144f2a9d3a266f888241f5 : SUCCESS in 1h 02m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f59aefdb53bb4a2489e4b43d1c758533 : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2173a706f74e4588861c1d672b14bd6e : SUCCESS in 54m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/86456aca2a6f49f5891cb64c4038018d : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"682e58462bb1347eff52206776ebe5cce552cf19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:41:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"12413d12462d176415a25fe2cc9244243db131a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 10:42:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2429bda48ac2443ebab44c6651ff3eff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391457fc756b4ff8b5202bd3efa88df3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dac11090b83dee3532cd1da39bf32d69824d45d6","revisions":{"14e42314f6e2aeb0e9073de8d04c792b4ca8b911":{"kind":"REWORK","_number":1,"created":"2023-03-20 05:21:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/877757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/877757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/877757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/877757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/877757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/877757/1"}}},"commit":{"parents":[{"commit":"22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5","subject":"CI: Retry docker push on publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 05:21:09.000000000","tz":0},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nIn addition to that, mark bifrost unbuildable since it\u0027s\nfailing on EPEL 9 enablement.\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14e42314f6e2aeb0e9073de8d04c792b4ca8b911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14e42314f6e2aeb0e9073de8d04c792b4ca8b911"}]},"branch":"refs/heads/stable/zed"},"dac11090b83dee3532cd1da39bf32d69824d45d6":{"kind":"REWORK","_number":2,"created":"2023-03-20 05:21:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/877757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/877757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/877757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/877757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/877757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/877757/2"}}},"commit":{"parents":[{"commit":"22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5","subject":"CI: Retry docker push on publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22518ed6c2660a80b8fdb3bb51d2c0cbe3eb8ef5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 05:21:29.000000000","tz":0},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nIn addition to that, mark bifrost unbuildable since it\u0027s\nfailing on EPEL 9 enablement.\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac11090b83dee3532cd1da39bf32d69824d45d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac11090b83dee3532cd1da39bf32d69824d45d6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
