)]}'
{"id":"openstack%2Fkolla~877760","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-13 06:43:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-13 08:00:43.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-04-13 08:00:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33","subject":"Test for unsafe files in tarfile.extractall","status":"MERGED","created":"2023-03-20 05:32:00.000000000","updated":"2023-04-13 08:01:42.000000000","submitted":"2023-04-13 08:00:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877760","cherry_pick_of_change":877759,"cherry_pick_of_patch_set":1,"meta_rev_id":"c71d98367508d4e3c006d048edab8afd60e00fdc","_number":877760,"virtual_id_number":877760,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 08:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-13 06:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-12 21:51:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-13 06:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 07:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-20 08:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 17:32:51.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-04-12 18:40:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2023-04-12 21:51:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"73b24f0d27eeaca2fcf163ab2bdd1a73960e2721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 05:32:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a0e297d50d231860d4dc92115b8c8a3621a98e07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 06:03:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c49b1ec2a8cd1b378cfd03841b2df5c29231ef9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 06:28:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e6908992b7e06b6477e02c9f470aec9fb8683146","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 07:55:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4281f32335da4c76ba1a11306fe36cab\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f8ef35f36824a9085dd2be07f61259f : SUCCESS in 29m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65e9b953aa574cdaa5b009bce412ea5f : SUCCESS in 1h 23m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b6d3dfa2ba040789e4e183414addb14 : FAILURE in 56m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/35730b2d3bc948e2951f84690b85512e : FAILURE in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6b5b635e3708d19bae7475a5252f9de0226fcae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 08:07:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eae0da332d6344fbb9162d6288245dde\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/243121cba6db4a73b6337b33705776f6 : SUCCESS in 31m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f07a883fa83c438fb92f6497b1cd0832 : SUCCESS in 35m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e5f4fe2ddbc541bca5a33b832e8dda2d : SUCCESS in 1h 11m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a421a7b2d21149ca98b837796b1ddca1 : SUCCESS in 1h 13m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8a83bd22cc5b4dcb9f75afa7c0d70046 : SUCCESS in 1h 30m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e49b18d1a2494797bafc2d7455d576d0 : SUCCESS in 21m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b7e308ec51144a0fa8a08472d1c0bb04 : SUCCESS in 19m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2dc035f5c15e470f86f419665fc12eae : SUCCESS in 58m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95bf05f1a4241d1b1c05a65dd1106e8 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f752d4de784475a9b9411a3ef5188e2 : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5c038a8c8934c138029ec058f8bf3a5 : FAILURE in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f98be00340aa42eebccf1ce0014aef8d : FAILURE in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a2b1bc737e4896823cb1b23fcbfb45 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3add0b78db8548c287e0515a18056b77 : SUCCESS in 19m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/381d1ee5e9d64b99b05df264a1de8613 : SUCCESS in 21m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/54827c4f307f4687894edc762c06e931 : SUCCESS in 53m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6002a91dd3d4f69bcf9f960abce406b : SUCCESS in 50m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5717238e0d274f5b89f0b742d9cbaa22 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d54125016bb4da29392465f049a3767d2bacdb07","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 16:28:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ba6685fdca8e4c651c199a2c35ce573879d90e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 16:31:56.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"915c82a0c684a3171b5143f5f4c5806210687e86","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-12 17:32:51.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3f26b8227705d84354f846db25ebaec77ffbf52e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:40:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc192ae2c5664ba696935df450d7873b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/20bd959bf4354e69bdf8a439edb261bc : SUCCESS in 25m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/815ec5b805f04145bc04f60f01759a92 : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c03c1e313ddf4d3292efd7407b9737b7 : SUCCESS in 1h 06m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b697ed3324242599bde39334d0549ce : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f6fab13ea224e0e6113ef7d603d24a66a788d3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:55:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edabdbd410c741cf805ef799bc276490\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/be971918daf94dc49ee9b294c053d244 : SUCCESS in 30m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0154ec6f08e44e9b844b32261f2b0c64 : SUCCESS in 34m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/373e61b44aea476dab6bbe946b45cdbc : SUCCESS in 1h 07m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1341dda71893416293b9e366f3d41119 : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1fcf82ebded745b48c7c854f98cc2219 : SUCCESS in 1h 18m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5bd6af95168e4278a0322f3858211d47 : SUCCESS in 19m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3245d3d9e41740dbb6ab080b13d23ca5 : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ecf43e830a3c40ee8bcbda358855f346 : SUCCESS in 52m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1ce2e985ad457ea70e958ccbff88c7 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced955a5381145a694368654a2bcc8c7 : FAILURE in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08218a2a56f84e7f82b09c252e78e720 : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5dea03c4c8341f0933da6c3974a2658 : FAILURE in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0268c016fdb4672bab4b4ad8c5da957 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/762429a6dbbb4e53bad7c2e020ff5543 : SUCCESS in 21m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/021c0fdcab284b7e93d67955aaf6b032 : SUCCESS in 21m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/632781dbf13a4b49982fb961a28baa4a : SUCCESS in 59m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a31a4c9629004854baaea34be12062e7 : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cbc3a718f496496188811e87ceb75eb0 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ab1363287bc6a7ec547235b79f1f1431505000fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 18:34:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1da14850eaa382d7786d50eece448c0c8a161f80","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 18:40:07.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1a7519fb9f950d1acf91db215941a8c142cf4512","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:36:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc7c9c4fac264666bf3358f6d1de0790\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/99587de3f3024c3c903367d99b679bec : SUCCESS in 22m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbf9b410f77a45fda01c5f83909e8284 : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5461257fdaa4988b2ab0f22c23950f8 : SUCCESS in 59m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/85c351d9b0764263a41d3e6831c08a11 : SUCCESS in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b6c79aaff115318ba03826d47611ff0c9d439d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 20:02:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2382ecb1c608434492824deedfd11e8b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb3b24e9bd2e404887c231e4ca2fdbdf : SUCCESS in 29m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8cb7c92fa67b4745838503eb483c3b2d : SUCCESS in 24m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/665aec5d8bfd49df9c663bd648fcf40c : SUCCESS in 1h 18m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6aba9fb62d4940488d9bf58678e7a12f : SUCCESS in 59m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9a5b973581bf45c3829628e8ff7b345f : SUCCESS in 1h 03m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37e70628ad5349a48be71581c0a3f755 : SUCCESS in 17m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cbb7340baa664e1aba3f7d5787568f2f : SUCCESS in 22m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c5d268fc818418fb690363e9e42e420 : SUCCESS in 56m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8863b02b93a49e5ad27da554a3b2085 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec811171b3c342a0940cf1f5bd1b9d6f : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd18d6f0df5241d5abb55b9b6ac96801 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/067c67d47da04fe98f4d0edac1a6c05f : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ceb186364f4a09a550103a5694422a : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afb48f7dd9904866ac4296fcfa72ec69 : SUCCESS in 43m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/77a1d8de35e0437bbf3c8f74d3c625b7 : SUCCESS in 16m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbc766e8ebc345a5939a8a6226ef7e55 : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d98aa037d01c4010b4e2d198f221975f : SUCCESS in 48m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07e06b51bf364797b7e536f64993217e : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":5},{"id":"4b016479d2c13612a5bc18a4ecca852cfe453c02","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 20:16:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2ecd1b7c4be1ff878aeeaadad3c2c018da5cbe01","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 21:51:39.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1317e9de51ff17e9d7f74467507be4d212fb4d4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-13 06:43:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"96776a55875430eea2e6261ce6f7cd02003d2bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 06:43:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ea368c07075a645ec94cabc39c19cb674d11cca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 08:00:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/997a4f14405c444c9a0bbce1f48a6315\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5c4887bf3bd44f28a95cc1a7a60240bd : SUCCESS in 37m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5d9b7379ea064c49a0780aded4d46bfe : SUCCESS in 1h 06m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f7faa186c33349c9b14456994fd8a847 : SUCCESS in 1h 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4d1d2af19f4402b9c7d6022218aa6d : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95b3bf9c16ab43b997c0e9e792dc5b08 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fef8dc7e9403439d8e407e7259b69d6a : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0841a719bd84176bd16005cf1178cd3 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29aa7fe36bec4ad8855e1ba4cabd085e : SUCCESS in 27m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c1f781d258a4d06ad288f30023b869b : SUCCESS in 59m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f5ef7d82cd14782bd0eb6dfa795be3d : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":5},{"id":"18d5088aa39139f5d15e54f3ff51207d35359a55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 08:00:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c71d98367508d4e3c006d048edab8afd60e00fdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 08:01:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1652005049d34c6d9bc5a407bf407c04\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfa6ed2c823435db4e3908d9450908d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9b350df4b3188183914f96b44b633fab3ad9afa3","revisions":{"83cbe11f17952d53b1f5fd4f149a932490101204":{"kind":"REWORK","_number":1,"created":"2023-03-20 05:32:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/877760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/877760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/877760/1"}}},"commit":{"parents":[{"commit":"e9f0a9aaea1616c62a4014957e06c3b61fd9c50c","subject":"Merge \"CI: Retry docker push on publish jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9f0a9aaea1616c62a4014957e06c3b61fd9c50c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 05:32:00.000000000","tz":0},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nIn addition to that, mark bifrost unbuildable since it\u0027s\nfailing on EPEL 9 enablement.\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83cbe11f17952d53b1f5fd4f149a932490101204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83cbe11f17952d53b1f5fd4f149a932490101204"}]},"branch":"refs/heads/stable/wallaby"},"681c712c9ba65e4438fd00d3904eca8775e70bf3":{"kind":"REWORK","_number":2,"created":"2023-03-20 06:03:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/877760/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/877760/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/877760/2"}}},"commit":{"parents":[{"commit":"e9f0a9aaea1616c62a4014957e06c3b61fd9c50c","subject":"Merge \"CI: Retry docker push on publish jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9f0a9aaea1616c62a4014957e06c3b61fd9c50c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 06:03:43.000000000","tz":0},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nIn addition to that, mark bifrost unbuildable since it\u0027s\nfailing on EPEL 9 enablement.\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681c712c9ba65e4438fd00d3904eca8775e70bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/681c712c9ba65e4438fd00d3904eca8775e70bf3"}]},"branch":"refs/heads/stable/wallaby"},"1ab0becb2be61e3bfaf46351bbeeb61106b2beb0":{"kind":"REWORK","_number":3,"created":"2023-03-20 06:28:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/877760/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/877760/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/877760/3"}}},"commit":{"parents":[{"commit":"e9f0a9aaea1616c62a4014957e06c3b61fd9c50c","subject":"Merge \"CI: Retry docker push on publish jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9f0a9aaea1616c62a4014957e06c3b61fd9c50c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 06:28:52.000000000","tz":0},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nIn addition to that, mark bifrost unbuildable since it\u0027s\nfailing on EPEL 9 enablement.\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ab0becb2be61e3bfaf46351bbeeb61106b2beb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ab0becb2be61e3bfaf46351bbeeb61106b2beb0"}]},"branch":"refs/heads/stable/wallaby"},"b65c68e37ed662b5a75ca7860a7db0b121683557":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-04-12 16:31:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/877760/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/877760/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/877760/4"}}},"commit":{"parents":[{"commit":"e9f0a9aaea1616c62a4014957e06c3b61fd9c50c","subject":"Merge \"CI: Retry docker push on publish jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9f0a9aaea1616c62a4014957e06c3b61fd9c50c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-12 16:31:55.000000000","tz":0},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b65c68e37ed662b5a75ca7860a7db0b121683557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b65c68e37ed662b5a75ca7860a7db0b121683557"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"9b350df4b3188183914f96b44b633fab3ad9afa3":{"kind":"REWORK","_number":5,"created":"2023-04-12 18:34:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/60/877760/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/877760/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/877760/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/877760/5"}}},"commit":{"parents":[{"commit":"e9f0a9aaea1616c62a4014957e06c3b61fd9c50c","subject":"Merge \"CI: Retry docker push on publish jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9f0a9aaea1616c62a4014957e06c3b61fd9c50c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-16 10:04:47.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-12 18:33:35.000000000","tz":180},"subject":"Test for unsafe files in tarfile.extractall","message":"Test for unsafe files in tarfile.extractall\n\nCloses-Bug: #1990432\n\nChange-Id: I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33\n(cherry picked from commit 3d008b7f5ec2a54d004e8e9370f303ef9dc7858b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b350df4b3188183914f96b44b633fab3ad9afa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b350df4b3188183914f96b44b633fab3ad9afa3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
