)]}'
{"id":"openstack%2Fkolla~877815","triplet_id":"openstack%2Fkolla~master~Iec4f243d6fe95dad4ea01225a651e86bd1a0428e","project":"openstack/kolla","branch":"master","topic":"bump-gnocchi","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-03-21 15:09:16.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-03-21 13:33:20.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-21 15:09:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iec4f243d6fe95dad4ea01225a651e86bd1a0428e","subject":"gnocchi: bump to 4.5.0","status":"MERGED","created":"2023-03-17 15:46:59.000000000","updated":"2023-03-21 15:10:25.000000000","submitted":"2023-03-21 15:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877815-bump-gnocchi","meta_rev_id":"4db42e6acea6051a667f02cf6ab2df9e2d0be50a","_number":877815,"virtual_id_number":877815,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-21 15:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-21 07:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-21 13:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-03-21 13:33:20.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-21 07:27:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 17:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 07:27:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-21 07:27:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-03-21 07:27:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"659634225047199003c71274b75977baa736c5b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-17 15:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1fe6c7435673484b7d164e0b23850c2ffcc1eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-17 15:47:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32e94d241f876da4bce8aa90a0dfdb27e26d5368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:23:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ecf94db311449ac87f75699b3fff324\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/065f04bfc17c4d13aedb575a1b56e936 : FAILURE in 42m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b98acaddf9ec45918eed6dea69c5d152 : SUCCESS in 1h 13m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aca1caa2d03244e48eddcf40713b8e72 : SUCCESS in 17m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/094d3f81aa4f47d9aad295cb99ad8f31 : SUCCESS in 1h 04m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f417cdd0a74d48cfa0c69380603ff06a : SUCCESS in 1h 01m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a87141aa3f45009db8e94a271ea3c0 : FAILURE in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a7e5ae4b7df41a49c49e0e1b20f00ea : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e32756296d984faab4d7d15478a68478 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4940a978d6413eb39215eee175b549 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23d1fbf1c5834ef7a13b4fec37aa0222 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60563aee53a4fd59896d6c067d2caf4 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b46d86c3dc634948900e41b7b4cdecb0 : FAILURE in 25m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b797783dbe6f42dcbe23cee3cfbfa644 : SUCCESS in 1h 09m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/428e3992580e47089fe37bd8b3f4556c : SUCCESS in 1h 19m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9b882afa38d54901a3549310e64703fd : SUCCESS in 23m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7d08616beab1450aabd35262022e6eee : SUCCESS in 49m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/412e8407a917489aa272247cc3c07b3e : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3658aa1f1e927bdf7115665b5a5f3373eba3988d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:27:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5690b67f917c4689ab0d1fea90bd089d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1fd594621e6943228c9faefcf1e59d95 : SUCCESS in 1h 19m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f070e211312c49b88bdc9e868e62b421 : SUCCESS in 1h 27m 08s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac4b8f3b01d146a788b804af5ad235b5 : SUCCESS in 1h 38m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1aa4dda2a844a2f8af99ef189992533 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f15f60505c914203a96fcd080075f4f8 : SUCCESS in 4m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5ed475c8c52f40a7987c245de45dcfe4 : FAILURE in 1h 27m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e5ad53d3bbc41779cb6d390414ad3d2 : FAILURE in 10m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0c72a2b5de8af45cf2730ba768ef67ea6ca4b11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-20 13:21:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"52b2bbe0b3de0b93c734a0673356b0556b96db03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 14:43:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cefbb284136414bbc411a4d286eaa60\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/62363fdb5be34243829d3aa594a0540b : SUCCESS in 21m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e6646581523406092ccaaa7017a3749 : SUCCESS in 21m 00s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e5bd7d7575794dfcbe3e7c92620d41de : SUCCESS in 1h 19m 12s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e9e0766b76e45b0820d5007a01fcfb4 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa72cf8d85e140babe008d1eb8728f0f : SUCCESS in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4910dcdd743b4106ba857b88a58db3c6 : FAILURE in 22m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/10ffaf0a823b46ce8dbb071e14b6e00f : FAILURE in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b910eb1f57152c7009792ae7534206a1890b96aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 14:53:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ce1a1e32f5849cbaf88c56226c0fb90\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/25ff7a06d91a4bfaac168a7b33489587 : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3781d4042e0041fa88548a9494247340 : SUCCESS in 1h 16m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25e5717e22104d8a9af0dbf0b66cccfb : SUCCESS in 20m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3ca17f8e438b426d85830e9b8089e19c : SUCCESS in 1h 03m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/43caa553e0e84a1ba3a19ce4865f46bf : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6529ad83bd974957b47592b854d60671 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b3c0e4985f4446197ae63c8ac4bfad1 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ff01d4bfc03423a8fd094bb4886cad9 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16a68aca2f86464dae07d54e571b9f69 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d1ab15f6284232b17924c714db50a9 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/924b1b1a207644db8b35daf299a08c1c : SUCCESS in 3m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8d4e1daea5ae4a8e8ffb82333b030817 : SUCCESS in 18m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/070a7be029244bc28203b48f8ee940fb : SUCCESS in 1h 08m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9b138d9d38374a098eec5f003494206b : SUCCESS in 1h 16m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/74bcc17650394d84ae7b10b748569077 : SUCCESS in 28m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a6e09140665c40d1b785f593a416c96b : SUCCESS in 54m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/def1153560ec4f42aaf7725b26729f59 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":3},{"id":"59d3288a2d78a10c2e855ab8cd96759b989bc63a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 07:27:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"caf416cfc540dcb1e3a9351cb3a7f85794f637c6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 07:27:42.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"ee1ce98214dc3da7f37bbff7151e2721b97fca1b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-21 13:33:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4643e6f6baeeed5ca356ea4981f7a79a9109faa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 13:33:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5c4d48b002d8c651880dd75dffeefc22815099a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:09:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/212ca12e63b14346a9d6ffb64eda4d1e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c882a43abf4d4552be57c6eca84d2115 : SUCCESS in 18m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff28e7a302d64b03acc421667dd04d52 : SUCCESS in 1h 03m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6433dd8ebfb24feea73e577d61e9a921 : SUCCESS in 1h 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f35018a62f324215acd916585e68b0fb : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2ea8ac472b74c14951e990ec4ff64ef : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/373c1006cd3141048bb59b3b5487c891 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b108ddd7a614902bb608512da1b40d4 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca0e951262f64210b2dbfd7cd7899a17 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b5bd79e04d884872947980e59a55b34f : SUCCESS in 18m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/54f73e7e2553483495cb62e3dec6ffab : SUCCESS in 1h 04m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e9a79874462a46c39f5eab242516834d : SUCCESS in 1h 18m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/08405bd283504d81be09b24ddde35393 : SUCCESS in 33m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/96d138139f554be88a879f9e5e704d82 : SUCCESS in 53m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3350bd89cab643d688f5e2886f422b5e : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":3},{"id":"f8c846a452d8432b20266001c63a5271cfb2f992","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:09:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4db42e6acea6051a667f02cf6ab2df9e2d0be50a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:10:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086a48885fd24d73a0a89ccc48b510d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5aec02680848f8804f7420fd34f5e6 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6ba20d8122840c2b63978f0a5a4fccd : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"516b7cd5788ebee8004c23013075935bb21fd3f7","revisions":{"5d071337c631870d6ea27155d01333d36d58e424":{"kind":"REWORK","_number":1,"created":"2023-03-17 15:46:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/877815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/877815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/877815/1"}}},"commit":{"parents":[{"commit":"12d431e3994023ce735a2352554a99a5cc9431c9","subject":"https: remove dhparams.pem before running gencerts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12d431e3994023ce735a2352554a99a5cc9431c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-17 15:46:47.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-17 15:46:47.000000000","tz":60},"subject":"gnocchi: bump to 4.5.0","message":"gnocchi: bump to 4.5.0\n\nChange-Id: Iec4f243d6fe95dad4ea01225a651e86bd1a0428e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d071337c631870d6ea27155d01333d36d58e424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d071337c631870d6ea27155d01333d36d58e424"}]},"branch":"refs/heads/master"},"4e8212a7a2a1df2e2f0e2296f169965374ea4305":{"kind":"REWORK","_number":2,"created":"2023-03-17 15:47:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/877815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/877815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/877815/2"}}},"commit":{"parents":[{"commit":"12d431e3994023ce735a2352554a99a5cc9431c9","subject":"https: remove dhparams.pem before running gencerts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12d431e3994023ce735a2352554a99a5cc9431c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-17 15:47:50.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-17 15:47:50.000000000","tz":60},"subject":"gnocchi: bump to 4.5.0","message":"gnocchi: bump to 4.5.0\n\nChange-Id: Iec4f243d6fe95dad4ea01225a651e86bd1a0428e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e8212a7a2a1df2e2f0e2296f169965374ea4305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e8212a7a2a1df2e2f0e2296f169965374ea4305"}]},"branch":"refs/heads/master"},"516b7cd5788ebee8004c23013075935bb21fd3f7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-20 13:21:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/877815/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/877815/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/877815/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/877815/3"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-17 15:47:50.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-20 13:21:16.000000000","tz":0},"subject":"gnocchi: bump to 4.5.0","message":"gnocchi: bump to 4.5.0\n\nChange-Id: Iec4f243d6fe95dad4ea01225a651e86bd1a0428e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/516b7cd5788ebee8004c23013075935bb21fd3f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/516b7cd5788ebee8004c23013075935bb21fd3f7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
