)]}'
{"id":"openstack%2Fkolla~877828","triplet_id":"openstack%2Fkolla~master~I3d264848cf1d78a56ac8bdde29bac7b67efe8b14","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-29 12:13:09.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-05-29 12:13:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3d264848cf1d78a56ac8bdde29bac7b67efe8b14","subject":"Drop Docker Hub as publish target","status":"ABANDONED","created":"2023-03-17 17:10:20.000000000","updated":"2024-05-29 12:13:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"efc4e961cd1bf5e576cbcae53447a22b5f84c574","_number":877828,"virtual_id_number":877828,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-20 05:27:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-03-20 05:15:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 17:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 17:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 01:06:24.000000000","updated_by":{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},"reviewer":{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},"state":"REVIEWER"},{"updated":"2024-01-20 23:09:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"3ddebb86aac08cfd325ee36793846f8ed3f2a487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 17:10:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbb254af6afaa844de44208d8b1ee0c7ba118f2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:15:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2ba354ec34465cb79cdc10e5d622b4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3c5db81fb364233b4007c0c311cc313 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36b8beb50d474b82887cfafca892e703 : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5637783208286eaf7537f0d1101b88adfe1ebfc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:20:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b98a3b80a8cd409ea1483574c6b1114d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a7b35ecf924a8683f2d9648ac72e92 : FAILURE in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/661567feed284bb185ece9cc974ccdc5 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/751e729872b642b08852e14d4faad441 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42ea66cce539482a9089aaa10a2d5d5c : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4c3a4baa844504aa38f5793e4f9b19 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c4581f29ec72bea9071db74776be82fccc09f9ca","author":{"_account_id":26264,"name":"Gregory Orange","email":"gregory.orange@pawsey.org.au","username":"gregoryo","status":"Pawsey Supercomputing Research Centre"},"date":"2023-03-20 01:06:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3d3809d2d286966c5de03f77362e8315d63b5582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 05:15:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"da735d51c410e6371363c2839d27287ed998893f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 05:19:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b5b32b6c05491eb7a1cc2aca58770c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95081af517ed45c39b7e73c770380e13 : SUCCESS in 2m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2f9fa2c6a0741d0a105e49b3a133d3b : SUCCESS in 2m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c54506eab1f5d09696504d4c19db3e13b00204a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 05:27:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/594e7386a3ca457aaf876c54b0183582\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2642f47b95b4b06912c73f104001f37 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de5a2189c9bc45fca9a3ee2946df721c : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db87a0e563d24c1485b01df0d087ef11 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1250d52500e14c7f98d2e319f6aa19dd : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96a6ddc3dff4943bbd72f50185585ea : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"efc4e961cd1bf5e576cbcae53447a22b5f84c574","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-29 12:13:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4d6209916895dfdc4acdf2a684ca7afcfb50835","revisions":{"5c4adc4d2803d3c98ea9fcac23fa8c00356ddede":{"kind":"REWORK","_number":1,"created":"2023-03-17 17:10:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/877828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/877828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/877828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/877828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/877828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/877828/1"}}},"commit":{"parents":[{"commit":"12d431e3994023ce735a2352554a99a5cc9431c9","subject":"https: remove dhparams.pem before running gencerts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12d431e3994023ce735a2352554a99a5cc9431c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 17:10:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 17:10:17.000000000","tz":60},"subject":"Drop Docker Hub as publish target","message":"Drop Docker Hub as publish target\n\nChange-Id: I3d264848cf1d78a56ac8bdde29bac7b67efe8b14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c4adc4d2803d3c98ea9fcac23fa8c00356ddede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c4adc4d2803d3c98ea9fcac23fa8c00356ddede"}]},"branch":"refs/heads/master"},"b4d6209916895dfdc4acdf2a684ca7afcfb50835":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-20 05:15:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/877828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/877828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/877828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/877828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/877828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/877828/2"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 17:10:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 05:15:52.000000000","tz":0},"subject":"Drop Docker Hub as publish target","message":"Drop Docker Hub as publish target\n\nChange-Id: I3d264848cf1d78a56ac8bdde29bac7b67efe8b14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4d6209916895dfdc4acdf2a684ca7afcfb50835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4d6209916895dfdc4acdf2a684ca7afcfb50835"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
