)]}'
{"id":"openstack%2Fkolla~877970","triplet_id":"openstack%2Fkolla~master~If48652b0bcc5603a587bb43b0dbe9b69e698e005","project":"openstack/kolla","branch":"master","topic":"for-debian-bookworm-upgrade","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-06-14 13:16:01.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-02 12:46:01.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"If48652b0bcc5603a587bb43b0dbe9b69e698e005","subject":"base: install adduser on Debian/Ubuntu","status":"ABANDONED","created":"2023-03-20 15:40:56.000000000","updated":"2023-06-14 13:16:01.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8801a36c8ccf21fa41d58a95438616af42f552f8","_number":877970,"virtual_id_number":877970,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-02 11:39:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2023-05-02 12:25:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 16:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-20 17:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 09:35:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"261bc3a2154cdebec4d7f99456ccb999fca62067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 15:40:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"503a2a05fb68a07b0ee6d635fcb8d756f86cf917","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 16:06:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"17af9ea657f7aed211118239dfd56e0d944eb578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 16:08:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1ccc48544166325518d137f910999c035e0963e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 16:35:09.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95375df784c41e581fa56067748a9c3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aedf99d9ec6b43bca4e43141e44be637 : SUCCESS in 21m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e2194d778a54e9c8879b5ab926bcb60 : FAILURE in 21m 12s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/28a14b71a24242bcb3d3cc648f80820c : FAILURE in 23m 15s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/608b251c6a2846659c5655ace2d613ec : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce000e926a734b9591edbe9b10586d7b : SUCCESS in 2m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cc6aeaa0cc3a40cdb3fc0f239e85d0e5 : FAILURE in 25m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/31f0957f48114629b6f8e743da9dd7f7 : FAILURE in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6cbf5853672ced5e0e41b86899f4ef23cea7984b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 16:50:42.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2fe27508c137550add1c27020ced7657d79c192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 17:25:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab2d463f9651483c888c14d00989689c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3db0fd29510f4eee9d7a465c5e9c54b0 : SUCCESS in 27m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a6f20f61b10a46cbb67daa84804434d0 : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b2ff96769476430cb25763113185ede2 : FAILURE in 17m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ec54c06825394fc5bf7a2f2d8e06c612 : FAILURE in 17m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bca3502a0fa34203b22f5e68048e0d38 : FAILURE in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d6eca60db44a0fb9f70640a9db846d : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b07331fd37a9414faa79e3e7dde7ae67 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/320241bc9e0d41048f318b3154a3208a : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/524d61304fce45dc8a63100504d6b913 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a996fc3b3eca45fea9237bb3cf2b7b37 : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71d8edc2ec504215bf743d311e9c8ced : SUCCESS in 18m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f70b35c33ade4c2699f0ec1495b54ce3 : SUCCESS in 57m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/066263d090664b5e9539959fc6f2b8f8 : SUCCESS in 56m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/91e83c5897834962a6ebf01fabf938b9 : SUCCESS in 20m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1c2782a4326f48b6a79f0e9c165cbb46 : SUCCESS in 50m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e63cc797e2b94e358a22b6ddaa683b0e : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2a1da673a0b1761faaaa1968ca9dc64d5ca5d0de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 08:59:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f3eedc7b444212367ca1439396a7b4be68883bad","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 08:59:32.000000000","message":"Topic set to for-debian-bookworm-upgrade","accounts_in_message":[],"_revision_number":3},{"id":"ae248fb55219b304451111cdd25e48ce526ba1a1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:11:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ffe0c9ae25c5a08626e7f69fc8e65d436da60060","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-02 09:35:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9389960e7ff36e2d38ad2e767c13306fa9f3e167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 09:50:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"48315a6c538dd4aad97e818b648fbf9d84a7dae6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 11:35:57.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f519be013e74bb4a3b529f5138f435d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bd40f3ae642d4d6fb9f3619417447659 : SUCCESS in 1h 26m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2346cab02fcc423abebae2921b06ceee : SUCCESS in 1h 12m 47s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee36d8c016c34af7b0513d5713e08348 : FAILURE in 1h 43m 29s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/425d325cf15540aa8061bc2c963f60c8 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f70a261a8ef4fdfaedbf4ed5185e5be : SUCCESS in 2m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7eee346310d44f079a63f34436278404 : SUCCESS in 27m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/449dd21085d9481d8c4847535350d589 : NODE_FAILURE Node request 300-0021110019 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1bcae514023d99f74bb4e81658c479ab9c0cb0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 11:39:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d8ad18bbdd4dc7bc6c66f6cdd9f32f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8936888f712d4e228f9d535f5a25b147 : SUCCESS in 28m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ae52f13aedd645c9989f0698c6b3d2d8 : SUCCESS in 1h 25m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4e2e8d50e5ca4d80b1fef9eac683d7b7 : SUCCESS in 19m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/68fd73728938451c92a1f76f079f9a47 : SUCCESS in 1h 03m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/452f60f812344d42a38d5f539657feca : SUCCESS in 1h 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b22da4df04f4f5f9e995bf645790b3d : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2fd8868b64140de9611407e55687d50 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddea4a4562214245891be61582df9b41 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1816bbfa8cd440b2bc6a36ea6d4e3930 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900bc96051b14fbbb8cb34069a4376d7 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/781eeffd7dae4323a068bbc12569b639 : SUCCESS in 21m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/de7225b8fe0a415d84a1374ecba33cc3 : SUCCESS in 1h 10m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3ce82d3e627b4ca1a25b3caf1dbe0ad6 : SUCCESS in 1h 33m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/736aeddb0ed941e1a47724104de7a925 : SUCCESS in 16m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ff65683892e24dd8a43b0f89b9524d93 : SUCCESS in 48m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d403c51ba20a41ea997c1e005ed054f7 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":4},{"id":"ea6a6dcdf1b5af65e1b6d196476fb77751c31af4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-02 12:25:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9256869b771e455cb0f07791cc8be6dfa848a2ab","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 12:37:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cadeef2e20cc305fd9e3640ec38f066b25908be0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-02 12:46:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8801a36c8ccf21fa41d58a95438616af42f552f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-06-14 13:16:01.000000000","message":"Abandoned\n\nmerged into other patch","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"20eb6d1ccb0e2957788d07d6d20a77e0e70689e0","revisions":{"1cabc4dab3e294eba203a118590ded63e0ab7587":{"kind":"REWORK","_number":1,"created":"2023-03-20 15:40:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/877970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/877970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/877970/1"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nChange-Id: If48652b0bcc5603a587bb43b0dbe9b69e698e005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cabc4dab3e294eba203a118590ded63e0ab7587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cabc4dab3e294eba203a118590ded63e0ab7587"}]},"branch":"refs/heads/master"},"27a77f683aa6e3908abbbbc3b99f21945c1e267b":{"kind":"REWORK","_number":2,"created":"2023-03-20 16:08:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/877970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/877970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/877970/2"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 16:07:21.000000000","tz":60},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nChange-Id: If48652b0bcc5603a587bb43b0dbe9b69e698e005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27a77f683aa6e3908abbbbc3b99f21945c1e267b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27a77f683aa6e3908abbbbc3b99f21945c1e267b"}]},"branch":"refs/heads/master"},"1aeb351d6f2410982fbca65044cb81be583d7344":{"kind":"REWORK","_number":3,"created":"2023-05-02 08:59:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/877970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/877970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/877970/3"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 08:58:29.000000000","tz":120},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nChange-Id: If48652b0bcc5603a587bb43b0dbe9b69e698e005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aeb351d6f2410982fbca65044cb81be583d7344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aeb351d6f2410982fbca65044cb81be583d7344"}]},"branch":"refs/heads/master"},"20eb6d1ccb0e2957788d07d6d20a77e0e70689e0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-02 09:50:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/70/877970/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/877970/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/877970/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/877970/4"}}},"commit":{"parents":[{"commit":"3be1b71948154c80f3ec247084b5049b520602b8","subject":"Merge \"opensearch: move to yum/apt repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3be1b71948154c80f3ec247084b5049b520602b8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 15:39:42.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2023-05-02 09:49:14.000000000","tz":120},"subject":"base: install adduser on Debian/Ubuntu","message":"base: install adduser on Debian/Ubuntu\n\nDebian:bookworm container does not have \u0027adduser\u0027 installed so\nwe need to do it before we touch /etc/adduser.conf file.\n\nChange-Id: If48652b0bcc5603a587bb43b0dbe9b69e698e005\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20eb6d1ccb0e2957788d07d6d20a77e0e70689e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20eb6d1ccb0e2957788d07d6d20a77e0e70689e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
