)]}'
{"id":"openstack%2Fkolla~877986","triplet_id":"openstack%2Fkolla~master~I7092b561d81c362ca24f84a3d753e777e5625ab3","project":"openstack/kolla","branch":"master","topic":"bookworm-preparation","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-04-18 15:52:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7092b561d81c362ca24f84a3d753e777e5625ab3","subject":"docker: use python3-venv to create virtual env","status":"MERGED","created":"2023-03-20 16:45:16.000000000","updated":"2023-04-18 15:53:12.000000000","submitted":"2023-04-18 15:52:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877986-bookworm-preparation","meta_rev_id":"df7bb8925157abe0b4c976f15e60f782d5627cbe","_number":877986,"virtual_id_number":877986,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 15:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-18 09:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-17 19:27:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-17 19:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-18 09:54:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 18:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 23:16:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-04-17 19:00:19.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-04-17 19:27:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-18 09:54:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1f8c3888b89cb950d4f26590003e1871e5e2d663","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 16:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0695426be1656a9da0778d4c200aa7e9fc58f50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 18:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c74816a6da64545b75baec6c38ba5a1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bda748271c1249bc81934ea25cc6fa06 : SUCCESS in 20m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/92c58707b5bc42e9b4c2ea32cd9ef52d : SUCCESS in 1h 15m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3351d0524b9440ee9105faccae0fbd55 : SUCCESS in 21m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22ee7b78cde64e7eb4255da19c5f8f4f : SUCCESS in 1h 01m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/437fa69332974766837f4e13eda8b298 : SUCCESS in 1h 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11acf2f229bd48c0a9d72323761db538 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39772a476de94e82aa9c7bd38eac80b1 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c3e0affe4684dbd8f81e2e21918fbbf : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63bbf4119b8b4154b83829cd74378638 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2644d07471b74b6fb92e6b2f717df183 : SUCCESS in 5m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fe4eab918b6d43a69fbfd32d14ce00e0 : SUCCESS in 18m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f9d25000b8a14ab59b4510f48272fdc3 : SUCCESS in 1h 08m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/edcb25fb27b440a6ad957e8e230701f3 : SUCCESS in 1h 13m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8bd42dc3cb5e472c92111e34c72fa7b7 : SUCCESS in 22m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1cccff63f9084372b5b60d83ed799fdd : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ea4925feb90f49eea0a70c902ace1096 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ba6eec593b834d8d844a10a357ce42b80a943e00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 18:44:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd6f0223f78647d79dd16e0df878e243\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4d19ada19c494e22a7aeeb91835d35ff : SUCCESS in 19m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc42d5b0a811493ca8b19e7c59c30ca9 : SUCCESS in 18m 17s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e339f947cf2476489269164acaef042 : SUCCESS in 1h 56m 55s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0eff33df7b3543d890e58307397afb9f : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ee0fffdeee6451d95bfb69b79418fb4 : SUCCESS in 4m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/40d853f316304c5d9a97aaa079a203a8 : FAILURE in 20m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e01dec54de5249c3b7addce8f12a938e : FAILURE in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8723008fbea860755a92f8678f7fa6b79c4ec4fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-27 14:22:08.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"07f6670031f0f7449b2c500b9f5e332e32f4696b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-17 19:00:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25564107895f140eaaec1ac5a83f450d38a14eb1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 19:27:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c118273b7ce1ff6c89c73d0c808f4f133f528558","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-18 09:54:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2787b472864611f2d444e6860bc065cbe59c5f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:54:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1c1e6c813a641a9d139fa485143aecf4097fb28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 11:14:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0a80c1dbaea47b1816a781e378e8844\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1ff2817ed99f4c6db65e8718955c04d6 : SUCCESS in 16m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6348f6404c9e4d8b996843a61cec6764 : SUCCESS in 1h 00m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1563d9923b194cd990ec5790039eb4bd : SUCCESS in 1h 09m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a837fa03ecb1439db69dc65bb39a4cf0 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b96d948df6ce4344938d9b5c06e79370 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/305f7110f74b411aabbb0b99d97e7e83 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8ed63b0bd4445f84acfc5baec2c950 : SUCCESS in 7m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ba76d6a8fcf54187a00bc3ff4997581c : FAILURE in 12m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/045a5b17150e48eea28587b5c42fd40f : FAILURE in 14m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b620a8c8b68d4f05a735cbcb381fd03c : FAILURE in 11m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4f5c5280793443d6936c9cdb52a5ce46 : SUCCESS in 39m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3390cce57817452c91b856819e8bf52c : SUCCESS in 1h 04m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/55408a4fd8fd4145816966052f8ef48b : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"83f3de654610c9a6280aed88d598cf98ecd3ca3e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-18 12:52:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"444af0491149ee166bd01269d5805338860dd8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 14:15:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27502e16d36c4ad59adf58a93658e67d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d7f980dc668c4054a938c931b073fcf4 : SUCCESS in 18m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a7d795fe7de948cfa76257c5db44d264 : SUCCESS in 1h 17m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e6883360d3944674b7a9ea68d0bff511 : SUCCESS in 19m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c00756200e214685bdafec142edb4048 : SUCCESS in 1h 00m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/72a84f5040ae4c63a115e2e9be980459 : SUCCESS in 1h 04m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68fa133aec574d6ca7bca96b9fb65793 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ef2a109a5f448e856e534b35846aef : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07b8dcfd44dc4ab29b93d60b6e5f3c78 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/945d669fa0964553b22588f5fd8f43f9 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed62fc546894adaa1297a43a28fb734 : SUCCESS in 4m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9bf68d3e3ec9475182913b2d90a952e6 : SUCCESS in 15m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aaeec70c687c4bb4816f421668cff0a4 : SUCCESS in 1h 13m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9abe227d41404a80bf427655690cd839 : SUCCESS in 1h 16m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b7024a8076024d1591806b4423b2ca36 : SUCCESS in 36m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5751f2bba6fe4e728b59f14cb62052c5 : SUCCESS in 55m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4de7bc4abb7e47dfaeb579ca03d3b5ba : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4c3298055defc5440b569466b8c9fc3d8e537983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 14:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1600ace9fd17a7e745af34b689bd55a8c341dc8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 15:16:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c79844100541f9bcafbdbe139bdb62\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4b1c261be5b54a57ad98c24ccee107ad : SUCCESS in 16m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86f90952ce77467fa8c563ac5de66bfb : SUCCESS in 15m 19s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/55a6206283c54b82b852a83a1d109be8 : SUCCESS in 2h 22m 05s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a43a790c07949b4a19bfcba27fc8840 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/140cbee63c3e4381801ad1e22151e498 : SUCCESS in 4m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3dfbe6ca72424d2f8722a87ad616fbc5 : SUCCESS in 19m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3190729459344f8dbf4cdc8b8443f150 : SUCCESS in 1h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a539da0a232c466323cd2fa52aa8808c512255db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 15:52:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a478a95239e40e1975809552e718176\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c48e13a5fe4f42179d4299c4ba326eee : SUCCESS in 14m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a7e89d5c613b4f04b7253fd2f67b66e2 : SUCCESS in 1h 29m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/65a36cde40a5436fb25bbaad7025e73b : SUCCESS in 1h 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd69092660f451999778670b619447e : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01e24f0ee0684ed6bab08b0d1b6151ab : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/271e91c131b64733879a40291e3d3a75 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e06c2d2895e43d49c633410a86c690d : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/92c9813534764390b3e8c7670bee6ada : SUCCESS in 15m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/10c74416b74d4ab981c35a9cf6ae45a5 : SUCCESS in 58m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3ab40fe9aad442e183461b880722f2af : SUCCESS in 1h 21m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0436059d5ad841a5ba031a379c750daf : SUCCESS in 29m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/32ac675c76dd4e6fa57d95f2f4b62863 : SUCCESS in 58m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ab133cb0e4a74ad2a230935da64930a2 : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e4658904cb7df9f8525596d64bc7e97fdc23c214","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 15:52:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"df7bb8925157abe0b4c976f15e60f782d5627cbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 15:53:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8ec5609d1c14521adf3d5f9ca44a023\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f410b7979954e7a9a4d402626596086 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca16f7ae7da77102db2c1d533c401094479fe1ae","revisions":{"ca16f7ae7da77102db2c1d533c401094479fe1ae":{"kind":"REWORK","_number":1,"created":"2023-03-20 16:45:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/86/877986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/877986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/877986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/877986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/877986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/877986/1"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 16:42:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 16:45:08.000000000","tz":60},"subject":"docker: use python3-venv to create virtual env","message":"docker: use python3-venv to create virtual env\n\nDebian \u0027bookworm\u0027 does not like when someone installs Python packages\ninto system directories. So instead of installing \u0027virtualenv\u0027 from pip\nwe move to use \u0027python3 -m venv\u0027 which we install from distribution\nrepositories (on Debian/Ubuntu as EL9 has it in base \u0027python3\u0027).\n\nChange-Id: I7092b561d81c362ca24f84a3d753e777e5625ab3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca16f7ae7da77102db2c1d533c401094479fe1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca16f7ae7da77102db2c1d533c401094479fe1ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
