)]}'
{"id":"openstack%2Fkolla~878007","triplet_id":"openstack%2Fkolla~master~I099b21872a12c2dccaba4065965d9227ee2e87bc","project":"openstack/kolla","branch":"master","topic":"pyzmq","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-03-27 14:23:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I099b21872a12c2dccaba4065965d9227ee2e87bc","subject":"neutron-mlnx-agent: pin pyzmq at 21.*","status":"MERGED","created":"2023-03-20 20:11:24.000000000","updated":"2023-05-02 12:24:08.000000000","submitted":"2023-03-27 14:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878007-pyzmq","meta_rev_id":"16fc2ed13386e5981adc13baddfd326ed0c23424","_number":878007,"virtual_id_number":878007,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 14:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-21 16:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-27 12:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-03-27 12:51:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 21:27:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-20 21:58:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 16:11:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-27 12:51:45.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"c50d1ae3ece736319e592f81e81e89e491986e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-20 20:11:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8088cf7e89ca6e1051f0a99936979f9d9cb1910a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 21:27:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe6319af49d4ad0815d3c3297bdff9e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/69dfbd55470044ac9bda23543f5a43a4 : SUCCESS in 18m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cbf8cc127f5a4fab98aeb8408ba523ce : SUCCESS in 16m 15s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90e15d5ee79e46c79b1488136724eb83 : SUCCESS in 1h 14m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a068029c467b45a9879bf9ec0dc425fb : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/abda3ab65873475bb3606537461564f6 : SUCCESS in 2m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eb719ede55374d35931983ced9207d66 : SUCCESS in 17m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e596ad4b8f8c4a638b736c52b89dcb38 : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbd8c1d07154131314d1aa2ad770a607ba4083ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 21:58:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67178858f68c4c25bf3cfc15ad5de067\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ee0609534d8446aeacb26389dd672d91 : SUCCESS in 20m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c85ce5616efc4d9f84cce557334d2ad2 : SUCCESS in 1h 18m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fc85e2cee5bc465997f19dfef939c05b : SUCCESS in 14m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/57ceb7a76b5c4954b3aed34fc5ca35aa : SUCCESS in 1h 02m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/999da5786b1b4d2bb72f08d8690d76d4 : SUCCESS in 1h 07m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b48682a3ce84e6b80fa4b3b2e259885 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab8c087329a849098686a0c59cdc2c0e : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1fab5aa0a954239a7ffd5ff92eb7dba : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7ee1015a88a440fb3d877b0b80a18eb : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047d77c7217046e5811dad9edc0f2481 : SUCCESS in 5m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f0fd70540554babab98cbc457e822a3 : SUCCESS in 22m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/40b2da9f3a8e4a458d7e2f597175cdad : SUCCESS in 1h 16m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/56dd22f9a1ca4ec696f747999fd4ae91 : SUCCESS in 1h 40m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f54c801958f947848ef61ad582fce134 : SUCCESS in 20m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d1a2bef3d8f84c90aec111d181cade9b : SUCCESS in 46m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/41ea8dd8920246d09650e85132453710 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c9835be2ba32b15083e1cd9deed998464aa7f8f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 16:11:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ffaba1e0a930dc3a58bd52a50e3356028d250d80","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-27 12:51:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f451cb2c70b00b24615c9328b66ab9c36413a47d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:52:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49a33386b8237c36f7fa4c92355390ad649b3dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 14:23:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3675d0b9c3114666adb3fdf8582976ec\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3c0ecf808014481ba6079f3fca4daf5 : SUCCESS in 18m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6af3eee5496448e4a00b2228f3a038e2 : SUCCESS in 1h 10m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/20a18df94e42415893a2a224c38c67eb : SUCCESS in 1h 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a181e9b5dd4205b05a574225453a66 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e955eca7d4004788bce77cc89f5e0969 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17d41e4ac5c444b78cf5579eff492feb : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4773696eab6441ceb87d235943b27657 : SUCCESS in 4m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cf470cfc4c3743358ae5879d191750d3 : SUCCESS in 20m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f29057e751814c91bc44c5c78e074b6a : SUCCESS in 1h 04m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fa6fd1b3478248499ac7a299bda10b84 : SUCCESS in 1h 24m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/98781a2d890142eb82a470c3a13b311a : SUCCESS in 20m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/01b8820b25db4b72b85e32f4594720b0 : SUCCESS in 1h 02m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/29f043f1ce2847f9904250a72252c608 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6c46ed6dd1878ff2e3790bd0bfe88f3b3f89bd58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 14:23:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"185055d9ba1e12210b0a96653032b15faecce7c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 14:25:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff67529835f469dbe601a7a3b7bc297\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae00df01b48c4862a706b371ecde0154 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"16fc2ed13386e5981adc13baddfd326ed0c23424","tag":"autogenerated:gerrit:revert","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 12:24:08.000000000","message":"Created a revert of this change as I589a5ae43ccf52104b8cbce327ab13248d3bbde0","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2d5307a442994d1a9348926f0664bb94ef96248","revisions":{"b2d5307a442994d1a9348926f0664bb94ef96248":{"kind":"REWORK","_number":1,"created":"2023-03-20 20:11:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/878007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/878007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/878007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/878007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/878007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/878007/1"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 20:10:17.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 20:11:18.000000000","tz":60},"subject":"neutron-mlnx-agent: pin pyzmq at 21.*","message":"neutron-mlnx-agent: pin pyzmq at 21.*\n\n20.* fails to build so let us try version where upstream fixed packaging\nissues.\n\nChange-Id: I099b21872a12c2dccaba4065965d9227ee2e87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2d5307a442994d1a9348926f0664bb94ef96248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2d5307a442994d1a9348926f0664bb94ef96248"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
