)]}'
{"id":"openstack%2Fkolla~878167","triplet_id":"openstack%2Fkolla~stable%2Fzed~I7d9c53297db1a9bd14e3e69b58867293014af69b","project":"openstack/kolla","branch":"stable/zed","topic":"bug/2012508","attention_set":{},"removed_from_attention_set":{"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2023-04-12 15:33:15.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-12 15:33:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d9c53297db1a9bd14e3e69b58867293014af69b","subject":"Cinder backup fails due to absent multipath tools","status":"MERGED","created":"2023-03-22 13:08:27.000000000","updated":"2023-04-12 15:34:21.000000000","submitted":"2023-04-12 15:33:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878167-bug/2012508","cherry_pick_of_change":878217,"cherry_pick_of_patch_set":1,"meta_rev_id":"071960a9ed2fa29c60b6cafd9a9d5cdcd3a97f48","_number":878167,"virtual_id_number":878167,"owner":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 15:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-07 13:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-03-27 15:51:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-12 13:57:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-04-12 13:57:34.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 14:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 14:43:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-07 13:26:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-12 13:57:34.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"1d5b7091cf806490a1f17ac0afe10b96cdb126e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-03-22 13:08:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4a7270ddbb4fade3482a489e0225697b45c21b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:32:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62b9721a42fc4eddadd2a4de25c96b23\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/28aabb42af1f42979f6a0d5e9614556e : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b85385b68d344020aeb7db0d88799eb7 : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/19c9ede9e67f41f1ba43d1f50f25acc8 : SUCCESS in 23m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2108358bacb04715bbd3ec09ae7eb6c8 : SUCCESS in 50m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4397089f039248f78ab5e6d1050f8ea4 : SUCCESS in 1h 08m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736ec437a1604ca9a30baf4a05e6dc58 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da56dfc03d9e449b81cd1166b5806d61 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f1714c7ccfe441ba5b7e473dc9220a0 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c96b02aba874ec1839afe60ae846174 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/947430538c3b45d5b0668f8428106b70 : SUCCESS in 5m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8e85fd7d7d84fc69ef427fe606af9ad : SUCCESS in 24m 55s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c411394b739849b884fa1daafa1b7a3d : SUCCESS in 1h 06m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9a230b2d919e471bb2c58f677f1ed2d1 : SUCCESS in 16m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef3d34595fe2447fb68e8262ee8e6d91 : SUCCESS in 53m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d86ae36ac88240aaa4ba961f08f2b003 : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c0f8b6ff1400f2b95ded92511265b627535aaf50","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-22 14:43:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ecfc60adfec212985ab35b3aa87ac12c8c339d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 15:46:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cc4d25b82a74217924c7226d134584d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7e0f7e4a14a14d33b3c1ff89b6f1ece1 : SUCCESS in 1h 15m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4881ae6a0b3644adba1db00679357416 : SUCCESS in 17m 02s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8081665542e4a4f9dbb9826bb6aa7f7 : SUCCESS in 2h 35m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad9aafd84db44d1fa526db8953a7e1bb : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6623b6df1de44528b2713ad0b4896c4 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/721b3486559b40dead5f83e35e3baf96 : FAILURE in 16m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e36f4b6adee147a69c11a61008f0f42e : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"525c3e40e32565fce2ee445627f58d2912132d52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-27 15:47:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b74849724d18e18063796df2debfa306d6a0056f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-27 15:51:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9bb17bede44b552000943631dcb7092e3fce7d58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 17:01:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7e228146034579a95b5da1625c4a2c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a5a425b8b4064befbfa5964112c8642c : SUCCESS in 22m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c004dfea9745488baffef31258146da5 : SUCCESS in 21m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bcb789c06dfd4318be3eb770100429d6 : SUCCESS in 1h 12m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92ce2a268d814164b84b6b421567fe34 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98e4d5d395f04731848b23143b7634dc : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0341a6760b944eb39b8605ab5e0290c3 : FAILURE in 1h 07m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e1a7c8a82d5744f8aa314fac41b0f7f1 : FAILURE in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b558062b11c23ceece7a10c47ad59aa61a4f7758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 17:13:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf1a64ee9bb496b86b4978038700819\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b3a90781764b4c8b8c09235001a984ae : SUCCESS in 21m 56s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3d026581d3e441c5a89f1e6430d54d95 : SUCCESS in 1h 00m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1043c72981414ca29d7a27f1c033eb2e : SUCCESS in 24m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7582ea8bb978464cb6439eb580d97ee8 : SUCCESS in 1h 04m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/eb316683c83742aeaeff7ae48fa23e78 : SUCCESS in 1h 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f511de928f74332962235ffcd3517c7 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13abcb9a370746d0b04a077bb6147945 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee3b405c75724aaa96c61c8a1d9ec357 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6f3bbb8879b4291bdd7b9ab7ded8f28 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260ca5a8762d45398c47038a69b41fd7 : SUCCESS in 4m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7194c9c6e4d84183a177931a81bb056b : SUCCESS in 25m 27s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/14fc1dbf121147e682d3ff4200b44375 : SUCCESS in 55m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3193a3b67edc46938aa514ffaf354ca0 : SUCCESS in 15m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/da04cb5c54c947a1a6f6ed224781520a : SUCCESS in 53m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb70f524534543d48c33fde79d877755 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a877cefd8e1de350dc5d180d17ca5d58a9410a35","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-07 13:26:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9e562ce7c5c7499a7664b815b586b5feb61e587c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-12 13:57:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e3a2f586626d2252e25b5e3e025771d8e5e54da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 13:58:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b35e14b2dec547800563fa32e91fddd1e419017","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:33:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53c25da030504c8b8396d2a7fce625be\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/79a6964393bb44fb994d022930312916 : SUCCESS in 23m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/750298f97e4f4f348fd519e31521fae6 : SUCCESS in 54m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ab589d745649479b9c951e3bbe946182 : SUCCESS in 56m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09c9fc43e2c5427c8d93cb73773467c4 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c87d6c3ea014a5aaa3a7da00980c252 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87024ebc756f4b8e8f230dca4670d3bf : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d12c1effee8491291bc2cf74bb44811 : SUCCESS in 4m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8c4f5d8305ba46beaac4d463e5469fe2 : SUCCESS in 21m 42s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3bec8a2a08da425683e2f06d632a87ec : SUCCESS in 1h 00m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40a5adaf21f64eb395c0e0532b372db3 : SUCCESS in 12m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7dd27f1bcead4189929ef1ee96c58adf : SUCCESS in 44m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8549ad0294744ea5a1b846090ff6e0cc : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b5a42b98f7ca49d5021cbbfee370494c097bf67d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:33:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"071960a9ed2fa29c60b6cafd9a9d5cdcd3a97f48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:34:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876b039a18be4ad5943379ed8fe03c8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c9334d124c4231ab7698e7239e9e63 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"885b27a679dde8175a27803dd71e12f68c70a0f8","revisions":{"3028f7ca69ed8f05d731c784dd65e991d94d733e":{"kind":"REWORK","_number":1,"created":"2023-03-22 13:08:27.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/67/878167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/878167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/878167/1"}}},"commit":{"parents":[{"commit":"dac11090b83dee3532cd1da39bf32d69824d45d6","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac11090b83dee3532cd1da39bf32d69824d45d6"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-03-22 12:27:01.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-03-22 13:08:27.000000000","tz":0},"subject":"Cinder backup fails due to absent multipath tools","message":"Cinder backup fails due to absent multipath tools\n\nCloses-Bug: #2012508\nChange-Id: I7d9c53297db1a9bd14e3e69b58867293014af69b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3028f7ca69ed8f05d731c784dd65e991d94d733e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3028f7ca69ed8f05d731c784dd65e991d94d733e"}]},"branch":"refs/heads/stable/zed"},"885b27a679dde8175a27803dd71e12f68c70a0f8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-27 15:47:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/878167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/878167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/878167/2"}}},"commit":{"parents":[{"commit":"dac11090b83dee3532cd1da39bf32d69824d45d6","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac11090b83dee3532cd1da39bf32d69824d45d6"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-03-22 12:27:01.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-27 15:46:15.000000000","tz":180},"subject":"Cinder backup fails due to absent multipath tools","message":"Cinder backup fails due to absent multipath tools\n\nCloses-Bug: #2012508\nChange-Id: I7d9c53297db1a9bd14e3e69b58867293014af69b\n(cherry picked from commit d98b48092571e2e7730d259cbd638f67ab3e19df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/885b27a679dde8175a27803dd71e12f68c70a0f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/885b27a679dde8175a27803dd71e12f68c70a0f8"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
