)]}'
{"id":"openstack%2Fkolla~878332","triplet_id":"openstack%2Fkolla~stable%2Fxena~I5cf944c509f23d430665de458cbbdb2fd31db7bb","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-27 13:45:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5cf944c509f23d430665de458cbbdb2fd31db7bb","subject":"neutron: Add iproute-tc","status":"MERGED","created":"2023-03-23 09:00:27.000000000","updated":"2023-03-27 13:46:05.000000000","submitted":"2023-03-27 13:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878332","cherry_pick_of_change":878331,"cherry_pick_of_patch_set":2,"meta_rev_id":"3ab0e0275ebd9ce8201b338af2ac420501a1c84c","_number":878332,"virtual_id_number":878332,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 13:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-27 12:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-23 15:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-27 12:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 09:00:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-03-23 10:22:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 15:22:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-03-27 12:28:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e2fd12d9f2078656d0ef4c8e6fd753381b0fdb0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-23 09:00:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"3b49ad41e302bedc800c2da6edfb77e14696bd28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 10:22:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/193ba81ca48246e3bcf61f030b1dd6ed\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ae93e9b50ecc45b9a71736a6c9cfc160 : SUCCESS in 32m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/67baed083d564cdfb44a4047765eb7e3 : SUCCESS in 27m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c14f60913dd44b4cbae7c50b93e9aff6 : SUCCESS in 1h 16m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c2b1a94ffc234fd4b9780067fb24f399 : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12d363c623bf49bca8fc708adb835df3 : SUCCESS in 16m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/33b1fe3b9a5646349425395f547b2782 : SUCCESS in 19m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/58cf4ec73b95499e9465b67a3dc02607 : SUCCESS in 58m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895bad13a36b412b8efa80835ef0e20c : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a01f32197f1349a589a173f5a68e674e : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4562701ddb014f6db9748974aea76d8b : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd6828ccec4b4bdc8e91a8cd5acf716e : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb34f606ee134ac5afd38cd0bb0d9f2a : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1a6c3f6b7f0456cbce3c17c0425e4be : SUCCESS in 19m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73c1e9148d984109bc5173e7d0c90a3b : SUCCESS in 22m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef6ca87914164e1d9b0a59cec161b123 : SUCCESS in 53m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ba2d921c43b49c8a08341967b9e0625 : SUCCESS in 52m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d5516783e3a4a5baac15087367e9623 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5441de001320b690c617fa8616e94d915ca44674","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 10:26:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e39c9389d2f498e902f4e3721aa47f6\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/99827030d74c48d2953a9d23b26d62c1 : SUCCESS in 36m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07af32a4a0584a2eb7dc69c05f429216 : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43651b4109a54371bf8caa4f608728b3 : FAILURE in 1h 23m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7805c1b3ad274f9d845a1c2b3322e9a2 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a6c709c5548419dacb4f6d234e804a4 : SUCCESS in 3m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/de4241a475be4228ac54565cdb879480 : FAILURE in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48bd1f2c59e53f699ff590ecb6124fd7a5ead9df","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-23 15:22:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e2db063d0ce587bbf2f5784e8f38fedf1e1b9f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-27 12:28:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0dbc6dc74e6be1a2f0c5e609ac142f391199cae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:29:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"258dcb39cc3ce904058efa6bbe34aabd04e1557a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:45:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a3617681a34e5d94873dd769666c72\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/61a4b01e749d4e01a27b7ec34d9b8b22 : SUCCESS in 30m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/613c578869394b7395a33cea7068dd72 : SUCCESS in 1h 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b6fb9f5b0be46b78d60e065ccd74826 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/537896cb697b4c9f85268daadac8c00e : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b0af5390a0d4a20bfa436e66687fbcb : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7dac7af6ed5483d8bbb2f7dbb976e0d : SUCCESS in 4m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a044fb8ba1447838cc5bf9a64989f44 : SUCCESS in 23m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfd08c710af240f1ba3c4bef40039edd : SUCCESS in 1h 03m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39ff61c4177a47819e2ff2268e015c89 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f413aec7c930e4ed8a7baf9c96e9c108891e12be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:45:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3ab0e0275ebd9ce8201b338af2ac420501a1c84c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:46:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd51de9d894a449e97e0e06d9b273ec0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e7434d386c458bad1a62cf050ae9ae : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a86a7bf071ffa31feb3d0eaf35be31faeb20008","revisions":{"9a86a7bf071ffa31feb3d0eaf35be31faeb20008":{"kind":"REWORK","_number":1,"created":"2023-03-23 09:00:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/878332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/878332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/878332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/878332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/878332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/878332/1"}}},"commit":{"parents":[{"commit":"6b86b85a19c9f2fd0f5bc2ebb55a45a1e85234d0","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b86b85a19c9f2fd0f5bc2ebb55a45a1e85234d0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 10:27:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-03-23 09:00:27.000000000","tz":0},"subject":"neutron: Add iproute-tc","message":"neutron: Add iproute-tc\n\nCloses-Bug: #2000782\nChange-Id: I5cf944c509f23d430665de458cbbdb2fd31db7bb\n(cherry picked from commit d7d43bed84dff0c394a28730fef593bdfbb064f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a86a7bf071ffa31feb3d0eaf35be31faeb20008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a86a7bf071ffa31feb3d0eaf35be31faeb20008"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
