)]}'
{"id":"openstack%2Fkolla~878333","triplet_id":"openstack%2Fkolla~stable%2Fzed~Icbf54154e4a2dffd840cfda9b6a863247d28c226","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-06 16:01:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbf54154e4a2dffd840cfda9b6a863247d28c226","subject":"base: make curl retry on all errors","status":"MERGED","created":"2023-03-23 09:00:45.000000000","updated":"2023-04-06 16:02:37.000000000","submitted":"2023-04-06 16:01:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878333","cherry_pick_of_change":874723,"cherry_pick_of_patch_set":2,"meta_rev_id":"fea084099237f570731861cc9c2337bf07483011","_number":878333,"virtual_id_number":878333,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 16:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-04-06 14:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-04-06 14:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-04-06 14:29:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-04-06 14:19:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 09:00:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-03-23 10:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-23 10:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-06 14:19:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-04-06 14:29:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"36d060d5d4335b527a33f19b31cdc7a7828757e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-23 09:00:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"53de5055319ac9003cf3d94e77cefd1b3edd4586","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 10:12:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a4cf1c85b14d68822d0045dc55312f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ce1e46444c8c414fb0b1e46ff507f5b2 : SUCCESS in 35m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ffed7b8e7b3b44cf9870ff36ca75d68d : SUCCESS in 34m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c5ac2f0e5544c848f505d81bb97d5a3 : NODE_FAILURE Node request 300-0020811422 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18b0e578c30e4e78b756debd98a517f9 : NODE_FAILURE Node request 300-0020811423 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c228df8830c400f91012349e7448a6a : SUCCESS in 3m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/998db79613ba4c748c00ce94d102ba7d : FAILURE in 1h 07m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f98583af54a145539706081bdf22aab0 : NODE_FAILURE Node request 300-0020811426 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a84ceb37e4366458ab79425b4aed5d2a59176d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 10:58:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57fb2fabdfa64306abeecbe5736a79fa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4d97f179d8fd454d96d69296e2de13a7 : SUCCESS in 39m 46s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/031dad316531432cb09a8641d899a72d : SUCCESS in 1h 37m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cdcc0fe65cfb41c3b333d33a8736680a : FAILURE in 44m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4873731b853347e699c77c602f0ef355 : SUCCESS in 55m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/06bc8ae6b8b44a6a8ae20b8828cdc94a : FAILURE in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc73ab256b049ed955c869f00b308b1 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ff9debcd9f47128737cc1acad4f421 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a5becd92a448988aea08e554447aaf : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5cd39cc3a8c462292d23dd0343c29e9 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20a26f1b8fd490aaf64b666c179cd5e : SUCCESS in 4m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c51c2b04c4f04c908a7e49706ff94440 : FAILURE in 1h 04m 48s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/17f615e8951e4f409f24bf76b9dc9df4 : FAILURE in 52m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/71bb9fcd59334603b510f573f62af65b : SUCCESS in 14m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1badfe141cdf42d8afdc722b51b203aa : SUCCESS in 52m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cdaebebc432446cabec1b347f8e5e8aa : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":1},{"id":"41e7226a114873890fe2e992a2c25cfc3e8e9be3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-27 12:29:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"458c7796e00c46089bf5e22d072eca66dff48bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:51:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca31bf6dc614da58f9c46f72acced78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d641e8662d944e03829252ff15419013 : SUCCESS in 24m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b06443be00d549ce813e09525c159949 : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5dc8c31916634db19af6b209abd78cb3 : SUCCESS in 29m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3205a1b8e7fa4ba289229131e05b838f : SUCCESS in 54m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8257c8320f88489b8895e00d79137cc8 : SUCCESS in 1h 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b85e9a630b4da3a42e6454eca1a953 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6a371814b046e2b719ff19ac5a0b1f : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1427c56dd4fc491d9e8dbfd552698937 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15ca97446c534d9db6fb4036ffd66cbc : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976ae7a24c7b473b9aa4202703edd91a : SUCCESS in 5m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a685d6f577cb4f858222b6fd96007df8 : SUCCESS in 16m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/137b2f8a86454c83ac51e3006ed566a8 : SUCCESS in 57m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/36de9c37ce744186b3453a720716c743 : SUCCESS in 16m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/57d1fbe530d94fb7a9d2764164d7866c : SUCCESS in 44m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/754b2d3ed3224db79941364a43def359 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9dce3c3878d338829de80aa7e3e454d7b472e00a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 14:28:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5387986651964bb9aa610900a7b63d1c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f848986287c94596af43634d4d92f2f8 : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ca4e376574545e28fde93e5618a9fab : SUCCESS in 17m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/564e839917824bb5a2b591b9c884f946 : SUCCESS in 1h 56m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4014fc147175409c98e85a5d0d7cfd05 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ced14deb37f40c592ffc915ff7541b0 : SUCCESS in 5m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3c32e8960ee64a039f85a50343728763 : FAILURE in 17m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/606fe70cadef43fabac86cd1850de86c : FAILURE in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8b55ca9e5f858bd981a8d52559679e0058c04b1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-06 14:19:32.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b2a02f479d61248630486e90797ae2858dc25b60","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-06 14:29:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad6d468ac472f471e7c8b23a6bf39a8ebb3859b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:30:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1c52fd51147e141eb07345e5ce561565f53b3c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:01:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38800f0c12304770b233f5496a2a6661\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b31dab5273b14dc39fc9e73e82b508b9 : SUCCESS in 27m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ba9dad06d4e4a8abbb0fc36b6d28086 : SUCCESS in 1h 10m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b023ab161c2549079dee70fef80879ab : SUCCESS in 1h 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e61334c4a34b2c8a22936b0a9cc6ae : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b38bcfadc364b6ab6ed127702b36f9e : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/967c9321fce3465b99c230917adc0cea : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87674a4d369e41aa90ec70b98651d37a : SUCCESS in 6m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13bf816ca57c49d9ab2b6220240ebccc : SUCCESS in 27m 19s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8cdab2721363460bb40c1996a33d560c : SUCCESS in 1h 09m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d8fd2e70875f491b9a76b16a5d739cf1 : SUCCESS in 25m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc8fd8c2e4724503b49799cf5d1f9ad6 : SUCCESS in 52m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b5868f8a9cb4cbe93bd794aa8de5198 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1293f52302cef34e67ec11cbd78731bce773f758","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:01:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fea084099237f570731861cc9c2337bf07483011","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:02:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f63dcf94e54416182f231db49ea5697\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1262c257e64c4d34b44b15bc92384d5e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2459c93eedffa31eab82f9c2ce47dab58aa654a1","revisions":{"2459c93eedffa31eab82f9c2ce47dab58aa654a1":{"kind":"REWORK","_number":1,"created":"2023-03-23 09:00:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/878333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/878333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/878333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/878333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/878333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/878333/1"}}},"commit":{"parents":[{"commit":"dac11090b83dee3532cd1da39bf32d69824d45d6","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dac11090b83dee3532cd1da39bf32d69824d45d6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:00:57.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-03-23 09:00:45.000000000","tz":0},"subject":"base: make curl retry on all errors","message":"base: make curl retry on all errors\n\nChange-Id: Icbf54154e4a2dffd840cfda9b6a863247d28c226\n(cherry picked from commit 90e157e3e35798dc9b524469a1bd7df0602f0355)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2459c93eedffa31eab82f9c2ce47dab58aa654a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2459c93eedffa31eab82f9c2ce47dab58aa654a1"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
