)]}'
{"id":"openstack%2Fkolla~878335","triplet_id":"openstack%2Fkolla~stable%2Fxena~Icbf54154e4a2dffd840cfda9b6a863247d28c226","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-06 16:01:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbf54154e4a2dffd840cfda9b6a863247d28c226","subject":"base: make curl retry on all errors","status":"MERGED","created":"2023-03-23 09:01:01.000000000","updated":"2023-04-06 16:02:32.000000000","submitted":"2023-04-06 16:01:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878335","cherry_pick_of_change":874723,"cherry_pick_of_patch_set":2,"meta_rev_id":"74fce9867df894d58af0b90fc4631e871f324c16","_number":878335,"virtual_id_number":878335,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 16:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-04-06 14:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-04-06 13:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-04-06 14:20:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 09:01:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-03-23 11:07:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-23 12:11:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-06 13:21:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-06 14:20:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e925e60faaf0dc679a46cdf21d5c78a515a20722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-23 09:01:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"299bd6e6ca529e774728bf93fedcb8557ff6c95b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 11:07:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4781fe643c9b48fd94366c096511e512\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef997535b1534ab1ab5469b446977bd0 : SUCCESS in 1h 49m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc9d5d18bda549cea6d1cc2a38e91172 : NODE_FAILURE Node request 300-0020811469 failed in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d65173b722334c199566473ff46fcb88 : SUCCESS in 1h 57m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd7f723f2efe430bb3332cf7cf3925ce : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d18cbd0032e406bb8436a85d9899e5b : SUCCESS in 5m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4c87df5e8674e0ab5a7eaa13bdd1c44 : FAILURE in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ff2b94b2159f190056a0895214b62232614cffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 12:11:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c12805c7250468baf764bb5431ee67a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7b365e08704c4e299d841627784c7f40 : SUCCESS in 34m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c00704aa817541de9db4c3a1f98e60ff : SUCCESS in 41m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bfa7c7db109f4c5d9f6cd42b6f7f412d : SUCCESS in 1h 42m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/adeb81a496f64f7583b39236c2c35708 : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ff87eb6b245243c0a6af2d4c39cf995d : TIMED_OUT in 3h 02m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b4305b7cdec742c89ef06ecb973c973c : SUCCESS in 15m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/02e9ebbd8f244362adc7d4b40eb557e5 : FAILURE in 40m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2ee15b89d245fbb7f76f68e9ce02be : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7ef092e8a694f41a66566eaec6f29cf : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e3cb98182f4cf5853c1f83b1c260f9 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c3d59a400a14e3784830a437a7d7553 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b5127b5c3f43d3ba8c6e00fc2d9506 : SUCCESS in 6m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f1e99e7835c430586e3b602b113715c : SUCCESS in 26m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aaade50bf7424fdd8b90d2f04033ff59 : SUCCESS in 19m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d5be4d5b4cd840c28fe51d2b94a9e805 : TIMED_OUT in 2h 01m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e45e09495c54058b0796127ae2f81c7 : SUCCESS in 59m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/735cdfeafcd4401ebe7ec03be3483f1f : TIMED_OUT in 2h 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4367c0d5890a1bebe1c400fb9bdcdd2000342314","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-27 12:29:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"343ca4d4ffddaeccd906fa01ef6a8f815a20ff09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:41:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/267e3c5310484a058b7258f07d45d09c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e8d4bbf77ab4daabce36b45a778df18 : SUCCESS in 23m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/598d2d475c9d4364aa51fc319eb7d8b0 : SUCCESS in 19m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e27608ee2704e1e8669185875605a4e : SUCCESS in 1h 10m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3e16361eb8d4b8c8bddebcbf7383c7f : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4f3dfe7698a49e79d8e3631b85cecfd : SUCCESS in 3m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/596cecfcb88b404b960fb45e1b2f2194 : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a95cfe0a428ad23d81018fb06ee9c13555bf4563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:58:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a79ecd43e1a4b97a88328e5a3db10ee\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6408b791e4804ae8bffdbb2c2924ff21 : SUCCESS in 32m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/22b30e60bb124c93b46d28d93436f129 : SUCCESS in 33m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/adb4c246f03b41f6a77e386ab339dcf6 : SUCCESS in 1h 12m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/26b564e6dcb94c5686fe9b9da09372dc : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3b3badf2316244c78e3861616996bbb7 : SUCCESS in 17m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/57321eede2d14462bab9cfb7adcfca95 : SUCCESS in 19m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ef045d31ad3d4baf8bbd81944357fdad : SUCCESS in 52m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d632c0f1e634ac18e0e48ecbf6d1990 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/173691aea20b44fa8b4ca75c95f0d49c : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81d16d3b4eda487996a0f327d0e052e2 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b458eb77c5924bb89fa98026bbd30430 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b835e7aa0254a758f5b157f977d8284 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9a7173e6ce64b50b1bc18ac87d1b388 : SUCCESS in 17m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e262be8298cb45f9bab2a27f06291f56 : SUCCESS in 19m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5168438564242bdb2d2f0ee7f56da50 : SUCCESS in 51m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80eedb95f3184677951267b029b9e553 : SUCCESS in 1h 00m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d1718a7ed9a4bc9898ef76dc3f8ed56 : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"480fd8b8a71a7927e888acce4f0088f2dfe97ba0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-06 13:21:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09b77655b8b48e7865f893e8d474126173eb8f02","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-06 14:20:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8c7876fb30b70475ac623aecdadfbc21d77ea0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:21:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"955f965dbe6f07b56154efc73c3375cf34712001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:01:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7733d20746944ccc91aa349c12cb250e\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8724ac17570042c6b535602393cdb8e1 : SUCCESS in 36m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8d8b871c4d94422891519bfc85d1200e : SUCCESS in 1h 26m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69688da5af334517a7807edaa0209c00 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2473aa3a9b504b7faf6f97923a1ae16b : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b6d0ab1127d4f7d93c796d2920dd0a0 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345f896dcb50495daa9005b90b36b9f5 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/26b7600f80de4700b300d7af37ed315b : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b8c10f8c03f4466bba772230c6c5fc9 : SUCCESS in 50m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b07bbcf10aa7464b8661aba1d8b5a59f : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"99c167c9cd5658566a7f31dbeebe84e93334e801","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:01:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"74fce9867df894d58af0b90fc4631e871f324c16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:02:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bed10eb1509c4b198fe6df2a7b287909\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7bac21d2d647d8ab7414415a3dc0f9 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6055f84c9d21d5483db9c1af605cf41bba0721a","revisions":{"b6055f84c9d21d5483db9c1af605cf41bba0721a":{"kind":"REWORK","_number":1,"created":"2023-03-23 09:01:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/35/878335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/878335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/878335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/878335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/878335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/878335/1"}}},"commit":{"parents":[{"commit":"6b86b85a19c9f2fd0f5bc2ebb55a45a1e85234d0","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b86b85a19c9f2fd0f5bc2ebb55a45a1e85234d0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 07:00:57.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-03-23 09:01:01.000000000","tz":0},"subject":"base: make curl retry on all errors","message":"base: make curl retry on all errors\n\nChange-Id: Icbf54154e4a2dffd840cfda9b6a863247d28c226\n(cherry picked from commit 90e157e3e35798dc9b524469a1bd7df0602f0355)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6055f84c9d21d5483db9c1af605cf41bba0721a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6055f84c9d21d5483db9c1af605cf41bba0721a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
