)]}'
{"id":"openstack%2Fkolla~878562","triplet_id":"openstack%2Fkolla~master~If0f9b4dd058a257d0653332d1b663e150c717304","project":"openstack/kolla","branch":"master","topic":"test-malicious-tarball","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-31 11:42:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-03-31 15:28:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0f9b4dd058a257d0653332d1b663e150c717304","subject":"Fix test malicious tarball fail","status":"MERGED","created":"2023-03-24 21:46:35.000000000","updated":"2023-04-01 11:53:03.000000000","submitted":"2023-03-31 15:28:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878562-test-malicious-tarball","meta_rev_id":"18f1f4b8c7c7ea1279654a2f2112ba4dba9b6589","_number":878562,"virtual_id_number":878562,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 15:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-31 12:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-31 12:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-31 12:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2023-04-01 11:53:03.000000000","post_submit":true,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-24 23:07:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-24 23:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-25 00:46:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-27 13:50:21.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2023-03-30 13:53:08.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2023-03-31 11:31:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"97bef8a7adb56e150fe179392aec6d3bf444a388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-24 21:46:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b688f37caf9c4992f00862b20d75b7ffe615a3ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 23:07:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bfae68e9ba94517bbe9b4292611a113\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d23d954856d742cbaa13f0ae1a8382b7 : SUCCESS in 15m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07e8185c2061425b80577f9290a22294 : SUCCESS in 14m 24s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/792f274eafd04f39850790b55b6a7b9c : SUCCESS in 1h 19m 43s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71a12d9f647941139bf085c6db391e54 : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/46050a8dd8564f1785eedacfd8fe1723 : SUCCESS in 2m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3eed93a239924dc98951859eae8cd1c4 : FAILURE in 17m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c96b867f09104af4a7b062b5b1e09b3c : FAILURE in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"015c56409be3a296b69e77fcc7720e1ecad3bb14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 23:21:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa9009bab3244873bbacd5376eadbfc5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a24791984b70445b8a6ec45375e018dc : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9018af9fa0fd41f7ab34b489fc4f76ae : SUCCESS in 1h 15m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0b338801691148f4b16fe9abbc80c1ef : SUCCESS in 16m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6e68da8fff9540958095ef076c591717 : SUCCESS in 56m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c4c3c746e69c4dafb386d91ebee34c48 : SUCCESS in 1h 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6b68c51a034be7a36674e6074ebaa8 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5ccfd17e685491ea602285fedcc29b3 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/147be7eded6f447ab840a26ec7c7ad9a : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b71f072a2ecb46abbd8754963906c169 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee737f4649f480f80d46d8d6f61a27c : SUCCESS in 5m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e40d9a7c0c674f3984e4a478ae9c5148 : SUCCESS in 21m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ea68a4d3a2c840a6bc7ef160dd47c788 : SUCCESS in 1h 09m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8bad2c44596a447ea2a8a66019da46e4 : SUCCESS in 1h 29m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7cda37728b2f4e208f1426313cc3383c : SUCCESS in 16m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8b785d8d14c546e692335b7d1047c668 : SUCCESS in 56m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/371f9fdf94d345cbb350a8a5fc98a596 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"32d88b67e2f39ffa04d8aaf868a16e896d3a0ba8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-27 06:10:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3c4f7553964adcedd0ff099b222e0d345d121b6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-27 08:52:05.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58de963e2ac6f9b79114ba98f85bd001a9877af3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-27 12:00:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a32a71b71840bea3e8e418f2283e9fdd572f5fb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-27 13:32:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f68d011748c03d163b9fcadcf1d16e442540632","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2023-03-27 13:50:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d390da538421141d8c6be6ecbbb86ed6aae2355f","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-03-30 13:53:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac409a0897632bf599c1968bc7b3a4cdfd215337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 11:17:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce196c667be07bae184ed6e97e471aa269a167eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 11:18:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bdb4e3e7a04e885f5c9510fe90fe8dee5c4f03f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 11:30:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"42cb009c9e84eb30f00da12c7d357ce1523f7547","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 11:31:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9333104cfdbb069ad3a10c0e7efe0bf8d3a32231","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 11:42:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4a5eb3fe8f081f6a5aaaad049a1371590a7477f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 12:35:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0b2acb38e2325399c1305f6697ae6a9dea98e8b4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 12:35:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2f46c4c18431731dff266411503114466e3d60f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 13:52:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d7f66c576640e3b69bc9c90f276986\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/53df773bede9457e919c5cb14d16b475 : SUCCESS in 26m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d753a76ec4a4c13863d8d3c5ce9a14f : SUCCESS in 26m 20s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69d2d3d24af24011a1344cf5bb4c619e : SUCCESS in 1h 15m 25s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7972876dcfb64e50af36d716ab394f5c : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c6b72f8b5544312a858cd21aee28998 : SUCCESS in 3m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/74250f7ec0524816b83395d74db6bb6e : SUCCESS in 1h 07m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aaf1677f9ca6490ab683315bee3564dc : SUCCESS in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ddc01d098eeb8dadebc51a68af0e1ef0360791f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 14:00:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a06e616c66a48f7b178cfe7d3986558\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/889917fe31ac4421b1c43c7c250da0e6 : SUCCESS in 18m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f0e838714dfa47cba8300dba3239fd3c : SUCCESS in 1h 20m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/730496e9756a49ec9300d9eb8002dd09 : SUCCESS in 22m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b867c9dfe2654affbe0b051a4854bd62 : SUCCESS in 1h 15m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7346d931edd442de936d5d42337aa1cd : SUCCESS in 1h 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5973a926cc6e431aa6653e6b3724353c : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57060a5c4173462da802a5c377d20aca : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8764787a79e64a64944e59fa6dcfcded : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94922b7bdc474e8e94ac6e7ba1eadd02 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4a1592c06c4faf83785c79e5f00857 : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8abb94667b5c4d759efe264982c18dc9 : SUCCESS in 17m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/333a104c3a8a44bfae13ac80c699f9c8 : SUCCESS in 1h 16m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fef66f63fda44afbb21918552e22aa9d : SUCCESS in 1h 18m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4ac439a335a4bf8b2b2af182c215e09 : SUCCESS in 18m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7101e5ce0d2c4337b3cbad47c4352c49 : SUCCESS in 53m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d88bbc05907c4462b49f22450f659cf7 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":4},{"id":"994d5b86fa29ae85e7e48e74f385387012576405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 14:01:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"81b52cc80685190194811150a65587e29ea6631b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:28:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f6e2f44b3574b1aa6289d82e4002317\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/98fec5d331144044806d67d8ec77d345 : SUCCESS in 15m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e4a14eaa644c44749067cdedac16050f : SUCCESS in 1h 03m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d94c4f252b364ab4a8f3a44b35d1730d : SUCCESS in 1h 23m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa46a29ca9cb42c7bc42e5bc86a62b87 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fbce05ba8d44a8a8f5d15951a226a1e : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cb2897c15a14ecabdf4b1d523f73825 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16271954f284e78a549c6d4e264ebb8 : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e110f427b9f0461394d439f1aa750add : SUCCESS in 21m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/37a2b3594b0245afa32d7e95a8e7602b : SUCCESS in 1h 14m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9a32540abf4b4c3db80744eb645ba8d4 : SUCCESS in 1h 23m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b2ddc9aaa743491b911b72765d18b268 : SUCCESS in 24m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a4666357533843cb97870de006e0748f : SUCCESS in 58m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/98aaca74ee2243deb9f4176cebc26b3a : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":4},{"id":"773f33b2a6822f54da5dd17d6853c0577c72a574","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:28:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"895687aac24d8cb352e7afa1de5622a04f1fd28c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:29:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cee6710d8f45bd9f0918d73e3efb3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997852e086904dad8e70d29e568aa72f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4},{"id":"18f1f4b8c7c7ea1279654a2f2112ba4dba9b6589","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-01 11:53:03.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"143765fb67221cc51f1dc56a41ac2b67dddc453f","revisions":{"c5f9d50694ac3da57cab5daadcbe2e328aee0b03":{"kind":"REWORK","_number":1,"created":"2023-03-24 21:46:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/62/878562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/878562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/878562/1"}}},"commit":{"parents":[{"commit":"d7d43bed84dff0c394a28730fef593bdfbb064f2","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d43bed84dff0c394a28730fef593bdfbb064f2"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-24 21:24:49.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-24 21:35:36.000000000","tz":180},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f9d50694ac3da57cab5daadcbe2e328aee0b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f9d50694ac3da57cab5daadcbe2e328aee0b03"}]},"branch":"refs/heads/master"},"3cdb2a2b61598f8a356a0bbc1466475752a893fa":{"kind":"REWORK","_number":2,"created":"2023-03-31 11:17:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/878562/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/878562/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/878562/2"}}},"commit":{"parents":[{"commit":"be15d6212f278027c257f9dd67e5b2719e9f730a","subject":"Merge \"keystone: Fix for missing libldap-common package for debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be15d6212f278027c257f9dd67e5b2719e9f730a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:16:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:17:46.000000000","tz":0},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nCo-Authored-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cdb2a2b61598f8a356a0bbc1466475752a893fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cdb2a2b61598f8a356a0bbc1466475752a893fa"}]},"branch":"refs/heads/master"},"48a269027dc5cca7b38f2cd1543ee5a51c2e3835":{"kind":"REWORK","_number":3,"created":"2023-03-31 11:30:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/878562/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/878562/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/878562/3"}}},"commit":{"parents":[{"commit":"be15d6212f278027c257f9dd67e5b2719e9f730a","subject":"Merge \"keystone: Fix for missing libldap-common package for debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be15d6212f278027c257f9dd67e5b2719e9f730a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:16:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:30:07.000000000","tz":0},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nCo-Authored-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48a269027dc5cca7b38f2cd1543ee5a51c2e3835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48a269027dc5cca7b38f2cd1543ee5a51c2e3835"}]},"branch":"refs/heads/master"},"143765fb67221cc51f1dc56a41ac2b67dddc453f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-31 12:35:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/878562/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/878562/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/878562/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/878562/4"}}},"commit":{"parents":[{"commit":"dda3bf45b29537f4925dc3374cf96eb8ed5b652b","subject":"rabbitmq: Pin to 3.11.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dda3bf45b29537f4925dc3374cf96eb8ed5b652b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:16:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 12:35:42.000000000","tz":0},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nCo-Authored-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/143765fb67221cc51f1dc56a41ac2b67dddc453f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/143765fb67221cc51f1dc56a41ac2b67dddc453f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
