)]}'
{"id":"openstack%2Fkolla~878649","triplet_id":"openstack%2Fkolla~master~I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-24 08:00:42.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-04-24 09:25:02.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-04-21 13:40:58.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-04-24 09:25:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3","subject":"docker: implement macros to alter upper-constraints.txt","status":"MERGED","created":"2023-03-27 14:56:37.000000000","updated":"2023-04-24 09:26:05.000000000","submitted":"2023-04-24 09:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878649","meta_rev_id":"cee78106875ccd1e4aeb720952090c8906b3b7d9","_number":878649,"virtual_id_number":878649,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 09:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-24 08:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-21 13:40:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-24 08:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 16:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-27 16:26:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 18:28:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-07 13:37:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-04-17 18:31:49.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"ba45f4447e113f2a5ea4261df31368645b92ae7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-27 14:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbac12e527ab60899b0c3732d495f8d60f0844e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 16:20:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41cf61d4576c42358c0abb9f83655e4d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/33623dea2a874a2da146c76a76b471c0 : SUCCESS in 1h 14m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1cc75842eacd4383a8bed66765b7de9c : SUCCESS in 1h 21m 29s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bfa3ef2821264a528697f4b376d67295 : SUCCESS in 1h 20m 10s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4a64e915c904cd29420b634d9929179 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3fa6b9ceb07b48febee72c1abeb55aa5 : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/22e5a340945142a5bceab13805db4fc6 : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/73e3208c94704b4d9b69b9ffee86437c : FAILURE in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d66b3161d5ef75c889dbc67a5570186b1f7530f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 16:26:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f287f4523ce42e48ae55ed8239d7506\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8641cd69ad904a6185ce988925716114 : SUCCESS in 23m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7b0cfa1d5e4040338a0acc6e45649ba4 : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4b406193fab04170b0b99d8e6cac926e : SUCCESS in 16m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cf00ec53eb5f4dc0bf4e98acc04e0923 : SUCCESS in 1h 07m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f907d4aeb4b04b9ba4d877e54a5b7fbd : SUCCESS in 1h 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad71d6942c14d3586d00f7e4ab5d10f : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/648b093c07d446f2beb5629f9cfff30d : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b12a7f4a31a543e6b843f3581b62c060 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e8e1eca33ed4eb99f642e57f74d29d7 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcfd0a8fb2df47ce9b7b59912b9286e8 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dadf05ad4f464dd295ed1c85b88a77ed : SUCCESS in 24m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/97b91a33077f4b278eefb7692c1d9ed2 : SUCCESS in 1h 13m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cff7360e19b542528af5df62113499c5 : SUCCESS in 1h 23m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/49cd8e22357b44bbb62187b2464b20a1 : SUCCESS in 14m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bd0662ccb94e4ceea8756d5b2d56b4d1 : SUCCESS in 59m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73eedcd9f1654eb791c293877f66dc20 : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d134a661d27862b799b50df7cc597967aa33f94c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-27 18:28:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"220be2eee988d07eb975887052a6059d4e05c022","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-27 20:34:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f39f580f20ac15709dbe4287285e28b8d7c3cfdf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-28 04:52:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0bb964a857d7cf5f039adeb9710082f5ac3cac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-28 08:07:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7e36fe6db39972d351e4823f8912cf5b40653d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-28 08:07:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"df1c679ef85e71704c15c85d589651faf1cfe2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 09:31:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2256c2b9d934491827360560eb1b93c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7336745a1fd54a20ae1f252799bb4852 : SUCCESS in 20m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2b6a2453d41f4a8f8548a04a23507ad8 : SUCCESS in 1h 20m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/40a3b6a0f47f4efc8f533452f5820600 : SUCCESS in 19m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/829c8867bd93424ab188477c2cc0ea2e : SUCCESS in 1h 16m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/22143dafb5e2455a9d3ce2a8025e5e99 : SUCCESS in 1h 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ffc527a5dc4fa28d57a458357b7f61 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f78f9aad49954cf29e1ca5c2bb01ef5d : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85e183c8d784413ba3513b7957112feb : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bba7bc3022544969aa09c737a52b503 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb2129d354847dc9910d1bd0750a446 : SUCCESS in 3m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1eb8077d567f4e54a0f92e952b4d8af3 : SUCCESS in 17m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0b4328260ec44b708a7c0b625c532f51 : SUCCESS in 1h 08m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/932b64bfd022472ca9699cf0c70f084e : SUCCESS in 1h 16m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/638eee5b075c4798948ab7701bbb150e : SUCCESS in 21m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/64d22dcb6be14ff08d3de7b2ccc6cc18 : SUCCESS in 47m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/60ed562024674a5d8e4d791367c960d5 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":3},{"id":"2662b302ff42a38672502c7967a7cafe153a5db7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 10:32:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/155da339e72d4e9594a0a931af69dee7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1d0c12c229d14e3da538ff6ebcd2b1ec : SUCCESS in 24m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dab51b09c6bb414fa6129bf6c85de45b : SUCCESS in 23m 57s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ae6e13c7a1546c392fdd1af939d4160 : SUCCESS in 2h 21m 37s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a397dd89772741f3ba8f44a292914823 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1f63f44c5cb40678f3aa0578cfd3688 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f9af3e3f0c414fb79169aabad4c81543 : SUCCESS in 26m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03ea09e3a4154d12a104a0476b1026ea : FAILURE in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0692674e7f7ca88aa62337180505acd7fd6af13c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-07 13:28:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a6711a143d9ff8b38198b8431fbc7425bef404c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-07 13:28:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c111189fc8a6b38b38f07eec46c81d281e97082d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-07 13:37:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a9dfe244824be4d3109fd3a51af33277e889328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 15:05:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b298234fbc4243afac622d25b96cc849\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/49419df6195d48b0a8ddd015c6edecd6 : FAILURE in 18m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a852d315ac5a407d9a0acabc15121a46 : FAILURE in 20m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c41fe56820b64c9eb62f05f6fee46f3e : SUCCESS in 16m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f7a929944f334a878f2a92887f476f52 : SUCCESS in 57m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aee6afe445aa4b06a8960a35be17d1ce : SUCCESS in 1h 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2994b1ed68434ff1a53f994d353363de : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9d09cca29f4a41bb89acdbbc54a5fc : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a985c5ae970e4c60a51b25fb03c77c20 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c01bcd163bd4c67bfab82c3bd357fcb : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4e946d89ea43eca33126b4e9081e16 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/251c79f54f4a4eccb4dc6cd84ae6560e : SUCCESS in 14m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/96102fad55344d10bf7fdbac7d745a4c : SUCCESS in 55m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5e617bb7c9b440f881b1ea0fb11d2be6 : SUCCESS in 1h 22m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/533eb0643d6340eb95e3c7064498573d : SUCCESS in 17m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7ad60ac5faf942fb9717e49b0a32e8f2 : SUCCESS in 47m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bc1828e7a4234773ab83d9f75bcc75eb : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e8fa97cf3ed5fcb718ad217f19e41ee0fe3b766e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 15:36:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/facd80dc737140b4a0e87cdcc08edc1f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b808ca09149544ab82018685c7e5be08 : FAILURE in 21m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f51e623d98b4d3fb5088d04fa4e7be4 : SUCCESS in 15m 37s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d67d50324034a8285db7d51db9b19ad : SUCCESS in 2h 05m 05s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2372b2499804973ae6b9505b2289f92 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/136d21e2844b4d83a09c3f1be69b7849 : SUCCESS in 6m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ce013764aa3848e8b46488a6fa4094e2 : SUCCESS in 18m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0ed2a35ae31746b0bf0478814ff13107 : SUCCESS in 1h 08m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"128923239d2c3d51ad8d35605d31aec84e3488d3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-07 15:40:54.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"48b51e829defd96ee71c6e606e792eb7be33008f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-12 13:31:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"06cbbcc6f877bdda94d265b93ca3759f5c74969b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:48:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d02b9d73db12404b9aa5018e0ad75be0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6c0ef9eddaae41b196f95cc29f93dcf5 : SUCCESS in 19m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/efbb0b61b18d4c2fa61c8d503dd45c83 : SUCCESS in 16m 22s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d3f3c2ae4b014daca74be065c1fb0eb2 : SUCCESS in 1h 13m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b7c0ce6eaa048b3a7dbb5ef7257334e : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9df195ee32674f4ca64138f1b60fd135 : SUCCESS in 4m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1a0f15a9aed64543829e6dfbf343bbf4 : FAILURE in 17m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c02bc8d6d4384ef68e8ac77aa6ad2956 : SUCCESS in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5fcc4446fb1907aec30dd8b6c6098ae5872c1844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:18:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ef044c2ad14657be2a55dbeea84063\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9bf08f6c7e2f4e09b9752ced8b2278c0 : SUCCESS in 15m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a31b4b971a6d4c94a60318ae21bbc06c : SUCCESS in 1h 07m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/501f493a777c45b28c881926c9202ad0 : SUCCESS in 19m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/80eccf55bb47416ba8f4acb52df7c9c4 : SUCCESS in 1h 13m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/caf7a7f84c7d4556a536da5d8feba532 : SUCCESS in 1h 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b084206656c648c0874b56a4bba592f5 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a280c3e48dd9428e8e598f0002448f4b : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5dccb22650e44c783fb2d2c190757d7 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88b44353c1794161bc79ae4c2cfdd48a : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0171ebde2bac412393cc5dbabdb59708 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71ad7ebd4b5145f9988b3b6b069fdb88 : SUCCESS in 19m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1eb5caa24caf43c793b7965938e2b0bb : SUCCESS in 1h 08m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b10426551088491697cc1864e23a518e : SUCCESS in 1h 31m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4c11d258e77f4b86b45e65d6811850d2 : SUCCESS in 18m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2893abfdf5f8445896a42a6ab571f876 : SUCCESS in 49m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bf8e1babc7f645deb1608619df53cbd5 : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":5},{"id":"96ca04fcbcd0196ba343a2f1c4bfe5296e6715c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 16:33:28.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"16bd41db62c6cf7a2b2abe09ab3e9658becf8a20","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-17 18:31:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7133f65dd3f097a511095a75bd11eafa20578066","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-18 08:53:59.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3eb4ed5da229243e2408414667a5b655499450c7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-18 09:47:00.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"14e4ad19e12f132ad6614880a61a545ef06b5cef","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-18 10:01:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"396fae5084a1b695101819b2ba2ede9ce2f286a4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-21 08:02:57.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cd1ba8c40efb09313249775177921cbac5107aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-21 08:03:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"6df3140091c825f4137f1f80cebefe17825b926c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-21 08:25:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9ba78e142e86754c903bf1da6a5ba7f27ccfc30b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-21 08:44:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"74625138dcc4314dddaaa08ec8b288cea62a224a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-21 08:55:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"649b2b58bad587b048c3a3c997e2cc335fbc99a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 10:21:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e436eeebba64598a8374bf2d831cc4f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/86907bc7e83b4ec2a8ef701a8ef276f6 : SUCCESS in 18m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/810ec3282790421aa3c1305dc68f143d : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1da142efd43545e68cc37db5501ef4f6 : SUCCESS in 16m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b7807766cbc645918ae49f93909a3885 : SUCCESS in 1h 01m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/eda06072e74e460493651b36d9255732 : SUCCESS in 1h 08m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f5f531065047a8bbe0f6a05fb3c465 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce057d41e70b41f695a675885d4c2b55 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a39ba75b5d24debb1993f10ad6aca94 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/188956142813401387bd889a1a8068b5 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5008c7eb4b0045f3b4702ad25e04708e : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7de9f42369c4e25adbc88ec8e51e5ca : FAILURE in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/45acb3b7ee314a30b5a5f3e8980be977 : SUCCESS in 13m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e50a3725477c455d8412fec86e8511f2 : SUCCESS in 1h 03m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c5902345b0d345879ac2504213bffb08 : SUCCESS in 1h 21m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5054c412ddde4a2b88b4a212d7909537 : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e557cea6e3e14ac7be496a0a45d502bc : SUCCESS in 52m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1514a80b079b48efa399aabe36e40e2e : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":7},{"id":"bcd3d1b907fa43f7383977cc104ea87a327e7e1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 10:48:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6d5d9e16996440785313071247346c2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/eba62b49fbca4213a84da679c169edfe : SUCCESS in 15m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/270545480d1c4495a6aad21bc09f953f : SUCCESS in 16m 51s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8cd9d37b13404a28adcf4a8078f7b17d : SUCCESS in 1h 50m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0048fbaf1b3e416a80720c9ffc416614 : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2158df8a12c645fd804894c8c2cc561d : SUCCESS in 2m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/64e94502ee1541078a06e319582f0e97 : SUCCESS in 20m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b93b692388b249aa9d2b76dc9cb68def : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f249aa514847e2df094ec36e9930ca789a1a393f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-21 12:13:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"41db8373c92be2b0ffeea5be3d51ede622076cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 13:36:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d50af6f7a64d2380b04bc7194c567d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7625b32275d8469593aa5af267e96b2a : SUCCESS in 23m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4038867d462c456fa556d48c0eadf317 : SUCCESS in 1h 20m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5a1f450c4c51413ab667af270601b6ff : SUCCESS in 17m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3b7713e26cfc47a49cfcf8e2065b0716 : SUCCESS in 53m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/83cc3dbf25a044f8bcb8cda6a7e28812 : SUCCESS in 1h 05m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb8b0efe0fd45d2b3d6836aff170626 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d8a039a3e3647f393e956f804dbba17 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ba3cf3e3bc744d6b9593eafcf306e96 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a54d46a14474bbfb748d42d8c8ff808 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8753264464c4caab9033f2e763c28df : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20c2591813c2486b9c622b1e9ebaef88 : SUCCESS in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2b45c15bb8dd4f13a77c9ab02c229f45 : SUCCESS in 19m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/11c6486fd82547d3bd89dd31f70f5af3 : SUCCESS in 1h 17m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ba2af9404994452592a06f41dd26331b : SUCCESS in 1h 18m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3d77c1bdefc14a53bab9879d048cd2c2 : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/36905ce18c6b4245ad4cdb003fa4e11a : SUCCESS in 51m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e57de1ccc3ca493e916914a1cc211c98 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":8},{"id":"b11d2a97526362ad303b09a828bea327245097f9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-21 13:40:58.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2ba45910b61e6278668f058f7af7ffb4d516c800","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 13:43:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36225288665f4c51ad3aa7dde14e20a5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/53cf19714fe64e97acc5f62c99e6169a : SUCCESS in 15m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c6ffe78441444f9797ea006bd52c98ec : SUCCESS in 15m 21s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c65432dac3d4e4a9ff81f37cd69f94e : SUCCESS in 1h 27m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1022f76e944d48f581b843ef93a83980 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66686499d4df436296038d1c1847b8cf : SUCCESS in 2m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/48105bea404641a7ac03cd8148400d2e : SUCCESS in 54m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a21f2cec38c4427295e4dc83f05b2182 : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"35b828a2789aff2030f32d6809a7cdee916ea064","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 08:00:42.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"be22a29d1e1220d9a0c27eef86b1c5bf7e290cc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 08:01:11.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2c5c86482b41ff2fddad759f84ac36f2642ba731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:25:02.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f47069b3b54235951bcea204404727\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6c7184eea2c949918079e04cec068d30 : SUCCESS in 23m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0cc0dcc393f7428c8544b163e943c934 : SUCCESS in 1h 06m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c2e6cd3f97c04efeb8552e826062494d : SUCCESS in 1h 04m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c36d588dcf4f7797e0b237525fb5ee : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e910c2b3c79c4efdab5a632a51440df9 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b55c890953b43ada3161f8933c6a1ec : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c1b21be98e45d58ec71c9ada312f78 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc3f24ce37144e0ca3cbfced5530cd83 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a5858b578e63423f89d2953943fb8af0 : SUCCESS in 18m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c275b4a491f343c3be736dba1dfa552b : SUCCESS in 1h 07m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/50cbfbebba014dcba74cef9cf6953565 : SUCCESS in 1h 16m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/93fd3fd1dc1f4228823e8b0c841093e7 : SUCCESS in 15m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b15f6305e7664026a213e4fd5759f6ec : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/27d6a5415af34efea8ef30897b0d0feb : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":8},{"id":"f9eb776f34554bfaba1a02a18c356f31d92a9bdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:25:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"cee78106875ccd1e4aeb720952090c8906b3b7d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:26:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c44c9ce1dba644b782c8d1735a522765\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e88a53be154053a801066c7dfae0eb : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49a3a39b15ab499d95f88f599a77fd0e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"19e40f160852a3519f3e58043b9b5555ca87f87d","revisions":{"db1151ebe6fcd689902921e14ecec936f66bd9d6":{"kind":"REWORK","_number":1,"created":"2023-03-27 14:56:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/1"}}},"commit":{"parents":[{"commit":"be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f","subject":"Merge \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constrains_remove() macro allows to remove line\nupper_constrains_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db1151ebe6fcd689902921e14ecec936f66bd9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db1151ebe6fcd689902921e14ecec936f66bd9d6"}]},"branch":"refs/heads/master"},"1d0412395aca3231d2ad1af75a77688980fded9f":{"kind":"REWORK","_number":2,"created":"2023-03-28 08:07:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/2"}}},"commit":{"parents":[{"commit":"be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f","subject":"Merge \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-28 08:07:35.000000000","tz":0},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constrains_remove() macro allows to remove line\nupper_constrains_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d0412395aca3231d2ad1af75a77688980fded9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d0412395aca3231d2ad1af75a77688980fded9f"}]},"branch":"refs/heads/master"},"c1ecaa811261a128534820394062247683c7f42f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-03-28 08:07:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/3"}}},"commit":{"parents":[{"commit":"be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f","subject":"Merge \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-28 08:07:52.000000000","tz":0},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constraints_remove() macro allows to remove line\nupper_constraints_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ecaa811261a128534820394062247683c7f42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ecaa811261a128534820394062247683c7f42f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3e2180c2ff017c98cfd713d8f3be6ef173780f9e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-07 13:28:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/878649/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/4"}}},"commit":{"parents":[{"commit":"bc2d3e4ef4aed12a5329f697967bd325339a7bff","subject":"Merge \"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2d3e4ef4aed12a5329f697967bd325339a7bff"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-07 13:28:17.000000000","tz":0},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constraints_remove() macro allows to remove line\nupper_constraints_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e2180c2ff017c98cfd713d8f3be6ef173780f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e2180c2ff017c98cfd713d8f3be6ef173780f9e"}]},"branch":"refs/heads/master","description":"Rebase"},"4ef4f8ee251acd6c2cf5f359815f21d82c28a156":{"kind":"REWORK","_number":5,"created":"2023-04-12 13:31:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/5"}}},"commit":{"parents":[{"commit":"bc2d3e4ef4aed12a5329f697967bd325339a7bff","subject":"Merge \"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2d3e4ef4aed12a5329f697967bd325339a7bff"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-12 13:31:08.000000000","tz":120},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constraints_remove() macro allows to remove line\nupper_constraints_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ef4f8ee251acd6c2cf5f359815f21d82c28a156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ef4f8ee251acd6c2cf5f359815f21d82c28a156"}]},"branch":"refs/heads/master"},"2a782058bcc86a3638f69dafa7dec9d53ae71fcf":{"kind":"REWORK","_number":6,"created":"2023-04-21 08:03:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/6"}}},"commit":{"parents":[{"commit":"2b0a5f684b80fcd85bfe12a51f899407e06741b8","subject":"Merge \"Update Stackalytics links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b0a5f684b80fcd85bfe12a51f899407e06741b8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-21 08:02:03.000000000","tz":120},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constraints_remove() macro allows to remove line\nupper_constraints_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a782058bcc86a3638f69dafa7dec9d53ae71fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a782058bcc86a3638f69dafa7dec9d53ae71fcf"}]},"branch":"refs/heads/master"},"7a399f3beaad94d88bb22c9480f980a405bb79fd":{"kind":"REWORK","_number":7,"created":"2023-04-21 08:55:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/7"}}},"commit":{"parents":[{"commit":"2b0a5f684b80fcd85bfe12a51f899407e06741b8","subject":"Merge \"Update Stackalytics links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b0a5f684b80fcd85bfe12a51f899407e06741b8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-21 08:55:47.000000000","tz":120},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constraints_remove() macro allows to remove line\nupper_constraints_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a399f3beaad94d88bb22c9480f980a405bb79fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a399f3beaad94d88bb22c9480f980a405bb79fd"}]},"branch":"refs/heads/master"},"19e40f160852a3519f3e58043b9b5555ca87f87d":{"kind":"REWORK","_number":8,"created":"2023-04-21 12:13:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/49/878649/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/878649/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/878649/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/878649/8"}}},"commit":{"parents":[{"commit":"2b0a5f684b80fcd85bfe12a51f899407e06741b8","subject":"Merge \"Update Stackalytics links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b0a5f684b80fcd85bfe12a51f899407e06741b8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-27 14:55:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-21 12:13:24.000000000","tz":120},"subject":"docker: implement macros to alter upper-constraints.txt","message":"docker: implement macros to alter upper-constraints.txt\n\nupper_constraints_remove() macro allows to remove line\nupper_constraints_version_change() allows to change versions\n\nThis way we have cleaner way to alter u-c in those images which need it.\n\nChange-Id: I8fc354b8aa4d03fcd3ecfb9cbfe75de67492a0e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19e40f160852a3519f3e58043b9b5555ca87f87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19e40f160852a3519f3e58043b9b5555ca87f87d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
