)]}'
{"id":"openstack%2Fkolla~878663","triplet_id":"openstack%2Fkolla~stable%2Fzed~I099b21872a12c2dccaba4065965d9227ee2e87bc","project":"openstack/kolla","branch":"stable/zed","topic":"pyzmq-stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-03 08:04:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I099b21872a12c2dccaba4065965d9227ee2e87bc","subject":"neutron-mlnx-agent: pin pyzmq at 21.*","status":"MERGED","created":"2023-03-28 14:27:10.000000000","updated":"2023-04-03 08:05:53.000000000","submitted":"2023-04-03 08:04:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878663-pyzmq-stable/zed","cherry_pick_of_change":878007,"cherry_pick_of_patch_set":1,"meta_rev_id":"dc3dee51469a7166c8d6208cb263708c8f790eab","_number":878663,"virtual_id_number":878663,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 08:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-03 06:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-03 06:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-03 06:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 14:27:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2023-03-28 14:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-28 15:43:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 16:05:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2023-04-03 06:35:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"c54460a995064f8b4ed34e0525f57f333d768862","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-28 14:27:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c58c8b662eb70bd35b32d63fa99a3ed5893ac112","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:46:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d25b9fca8c4a959af10a5ca1ee33bb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/71d835cd6f3d4405b3ab59a02344ee22 : NODE_FAILURE Node request 300-0020850002 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d61a062b68004424b1909fbf291bab37 : NODE_FAILURE Node request 300-0020850003 failed in 0s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0fdd5acf6c941639bcd870a453a3d4a : NODE_FAILURE Node request 300-0020850004 failed in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e6fdd34fa9a4210be863d33273e86a9 : NODE_FAILURE Node request 300-0020850005 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb0ffabebbf84f2aa651a9b6fcb68a24 : NODE_FAILURE Node request 300-0020850006 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/122c0b2386a241978ebd1ddfdb93c691 : NODE_FAILURE Node request 300-0020850007 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/52cf24560d644f2b92ae5fc5a30d98ad : NODE_FAILURE Node request 300-0020850008 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d83f117d286f4222d86bcca7dd6b2165e2f0d107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:43:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24aefcd9f6e04ef588944c82f63b0e81\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/669c3c2801cc484399e4b6ccdb20cb10 : SUCCESS in 49m 27s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d7cb8a3bd7c042069e59b447a58ad267 : SUCCESS in 1h 09m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f596d5bdce224efb83734ace12eefc42 : SUCCESS in 23m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/535e34fab73742f8bca99ebb33243d26 : SUCCESS in 53m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4cf42e8d0538467ba1484b8f9dd56ba2 : SUCCESS in 1h 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b58c101b032489da3dc2a0cfe27f38b : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc13677ed7a44abb54eb62cd6a0a340 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee1b52b8a3314b418197f629409bd7de : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a348ee1d26647b295c95887e3e3ee22 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a89037c4414616b8239fca03a33274 : SUCCESS in 5m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dadc505da28c47b39fc23644160ebea0 : SUCCESS in 25m 56s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/477af7b1bf9c45afabd9f64d8a3bff2f : SUCCESS in 59m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f1d4f171a414c7dad9d0b778e11b499 : SUCCESS in 15m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7ebd9ec19674715bbd547f190045afc : SUCCESS in 49m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/972b886b7a134b3a85581de028670f6b : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a2af9f07ea0c49b8c76fe31ba82d5a7e8b88befd","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-28 16:05:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc6915aa80197775883ba3b455be2a3edd5779ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 18:05:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3331ee99ff44f8abdf794c7ea6623b5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/24d8634d0d2549bca6f8fad5bbeacde1 : SUCCESS in 40m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f9284ea3d4f49da831e4f5c1ddf00e7 : SUCCESS in 36m 08s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/52dbac6dd9e04599be65476170e6380c : SUCCESS in 1h 58m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ddd113c031343caba11f715bd79b2eb : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/446cef442c5c4e29a19820d8056e505b : SUCCESS in 3m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b6f9c2d507224e95b25f2093d13acf84 : SUCCESS in 1h 01m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/124f6f5a6f3d4bfcbd8e1df5110b8185 : FAILURE in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61e4b0f9fa0142254d6c2f77f27dc5867fad29f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 06:16:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d22f6c957594c5863d03c414dff62364af10def1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-03 06:35:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a185124f486b1fb37bde2e577bcbdbb000844ee1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 06:36:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a535623453aa6f67451117bee6a3423f341f9ce3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 08:04:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6d61471d2846d9a7a03545deea7149\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b9910d01f5e644f7934b141c3fab482f : SUCCESS in 18m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/26a7d9c37975465b86e034b2b4522f7a : SUCCESS in 56m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/26879e3426de4adbaf0c4c42f284e178 : SUCCESS in 1h 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c614701c753e46b8a26b6ae0818ec2c4 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b8318b1ab249299fa7ee2fe4c9139e : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4681412f49943c383edf8990c6a996b : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa6283862f4437c863bf4de3c55fd0d : SUCCESS in 5m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dd8f3e917c3246b782cab301fd314337 : SUCCESS in 24m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8d3aef238e9e43abbc419771c6e545ff : SUCCESS in 1h 09m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/07e4ab81ea58472dbe851a548fc04adc : SUCCESS in 20m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/12f153b63b9346d090b03015ed9e8703 : SUCCESS in 49m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a4e65530b6a4ae4b40020865b5ed432 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"693670689aa01cd4a3c3153e178223e62b15e0f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 08:04:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc3dee51469a7166c8d6208cb263708c8f790eab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 08:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecf970db4fc047dc97e6851cb03e4453\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b7beb5ce484184944a8008a24a8d82 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"395e9e56594201ac06a5842eac32b0d65daafcf9","revisions":{"395e9e56594201ac06a5842eac32b0d65daafcf9":{"kind":"REWORK","_number":1,"created":"2023-03-28 14:27:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/878663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/878663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/878663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/878663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/878663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/878663/1"}}},"commit":{"parents":[{"commit":"a8d0098bd2575bbfab6155265d649bb0e4e729f6","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8d0098bd2575bbfab6155265d649bb0e4e729f6"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-03-20 20:10:17.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-28 14:27:10.000000000","tz":0},"subject":"neutron-mlnx-agent: pin pyzmq at 21.*","message":"neutron-mlnx-agent: pin pyzmq at 21.*\n\n20.* fails to build so let us try version where upstream fixed packaging\nissues.\n\nChange-Id: I099b21872a12c2dccaba4065965d9227ee2e87bc\n(cherry picked from commit b2d5307a442994d1a9348926f0664bb94ef96248)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/395e9e56594201ac06a5842eac32b0d65daafcf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/395e9e56594201ac06a5842eac32b0d65daafcf9"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
