)]}'
{"id":"openstack%2Fkolla~878665","triplet_id":"openstack%2Fkolla~stable%2Fzed~I66ea5711810ed0060cf89192a780bfcc92e79770","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-19 16:44:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66ea5711810ed0060cf89192a780bfcc92e79770","subject":"Replace invisible unicode chars","status":"MERGED","created":"2023-03-28 15:58:26.000000000","updated":"2023-04-19 16:46:42.000000000","submitted":"2023-04-19 16:44:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878665","cherry_pick_of_change":878767,"cherry_pick_of_patch_set":1,"meta_rev_id":"b0dd66d116c110d36d5079e51df5791ee5f5765a","_number":878665,"virtual_id_number":878665,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-19 16:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2023-03-28 17:20:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2023-04-07 13:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-19 15:19:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-19 15:19:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 15:58:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2023-03-28 17:20:39.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2023-03-28 17:43:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 15:19:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8c3f8a657de4249eb244817e063ed2ae3a87f108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-28 15:58:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3e351db17a345676a641517c215f778872f85d8d","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2023-03-28 17:20:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b273f2aebba3157b206e10f2c1c5831ac3385fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:43:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4240d6caa2334044be6d265f414e0c27\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/41622eec97594959929707b4a27bd010 : SUCCESS in 38m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/751f393cf9a047448c60400c042a1f10 : SUCCESS in 59m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c77e64aa1fd140d0aab59cd0b4086fbb : SUCCESS in 23m 11s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c1b6624470154bbdb26e28d23876dd47 : SUCCESS in 50m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6214787db9c745369cd226ee172d4d6e : SUCCESS in 1h 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd4458bede84cdebfbec333e4a369fc : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec16bc542d4e4e4a8b8896d0c2419f4e : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf1ad02beae54cd4beea67f1c59d12f6 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2d22ec8df3a43f6bc47d1c834bd382d : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b94205bb184926b388e6edeebab3a4 : SUCCESS in 4m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/62f14ced74694e8397beb52e7797bb3a : SUCCESS in 30m 11s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f05daefd306143b5bff53d757ed0c143 : SUCCESS in 1h 08m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0258483429624f63b7f940bc544472bf : SUCCESS in 18m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fbe9d992d534c1f86a3ad689ab92f4c : SUCCESS in 48m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7f6ede9e89a24a7f8b4ea3ebb5bcc0e9 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"492fe29f9e36f20735c5ce655251c235fe14231f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 18:14:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66c7191699724f3688e2e7426f58ebca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/83876e33bb4940d0ba121908b7d4f8f6 : SUCCESS in 40m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3be167bf6cfe48398034254c0b76b081 : SUCCESS in 1h 09m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea62112971b1445596327a61001d9d11 : SUCCESS in 2h 14m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d66c1b2f8a0b4f1685a368f8cef0711c : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5df42fe738d490694f2b182a0044754 : SUCCESS in 3m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ca3f88630e8e4556afe6053af56dc13c : FAILURE in 1h 10m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d48637b7e0404a07b4be713819450673 : FAILURE in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26735163e1ed5c2604756d5685d67b05e281381d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-07 13:26:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bbdf8deafe8108397c61c88d35ba9c8afff9dbc5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 15:19:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c3ee26223c944797d3e20dc2c2cd6033fd065f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 15:19:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"315502663a53efe3dd6a60f4cf3fd414b6b5889e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 16:44:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba55ad9392194be0904d7af0ab643537\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42e42556fe1a465d9c51fa41a5c8205d : SUCCESS in 17m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/561f189377284227872cf02fb2233077 : SUCCESS in 1h 03m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/78e1aee254fa477fab8b4a16b410d5ce : SUCCESS in 1h 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3943506f6e4925ad7d3588ae1f28bc : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4f9959a3e754726bed07799114312ae : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e28f38c546e4e64bda98ae7fc399294 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af5d806101e42c697b2be0953091376 : SUCCESS in 5m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/be9e1da50c474b26a5a589b9a050c197 : SUCCESS in 25m 35s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b072b3341fad483b8b72884f04d0b379 : SUCCESS in 1h 01m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4ab373cd4d7d4768b9515146fa3d0a3d : SUCCESS in 16m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13bd40c6ede04dc391220b2ceac808a0 : SUCCESS in 47m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc4c75469a96438fbd534ead64dbbe73 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"604c44dd34119ca62706fdf2950574b271071320","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 16:44:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b0dd66d116c110d36d5079e51df5791ee5f5765a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 16:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43723132bfc44f0ea82cc622d00ebea8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fce5b40e4cc45b4a08a74c27f8bd150 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cb59e45a592056b01b2f3a401535738f87de35c","revisions":{"2cb59e45a592056b01b2f3a401535738f87de35c":{"kind":"REWORK","_number":1,"created":"2023-03-28 15:58:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/878665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/878665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/878665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/878665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/878665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/878665/1"}}},"commit":{"parents":[{"commit":"a8d0098bd2575bbfab6155265d649bb0e4e729f6","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8d0098bd2575bbfab6155265d649bb0e4e729f6"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2023-03-28 12:08:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-28 15:58:26.000000000","tz":0},"subject":"Replace invisible unicode chars","message":"Replace invisible unicode chars\n\nThey\u0027re believed to have no special meaning and were put by accident.\n\nTrivialFix\n\nChange-Id: I66ea5711810ed0060cf89192a780bfcc92e79770\n(cherry picked from commit 2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cb59e45a592056b01b2f3a401535738f87de35c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cb59e45a592056b01b2f3a401535738f87de35c"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
