)]}'
{"id":"openstack%2Fkolla~878666","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I66ea5711810ed0060cf89192a780bfcc92e79770","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-19 17:11:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66ea5711810ed0060cf89192a780bfcc92e79770","subject":"Replace invisible unicode chars","status":"MERGED","created":"2023-03-28 15:58:35.000000000","updated":"2023-04-19 17:12:00.000000000","submitted":"2023-04-19 17:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878666","cherry_pick_of_change":878665,"cherry_pick_of_patch_set":1,"meta_rev_id":"2c00c31bad6de7217f6cc9d966f5a6f88330b5a4","_number":878666,"virtual_id_number":878666,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-19 17:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2023-03-28 17:20:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2023-04-07 13:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-19 15:19:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-19 15:19:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 15:58:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2023-03-28 17:20:52.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2023-03-28 17:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 15:19:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"7335de39acd19ca1fff998725c49238ef613b75b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-28 15:58:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"64d09e9c0af1c463e0aed0c69cb3466099567971","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2023-03-28 17:20:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a67a41684b85ded51ebcb7d5db06d28791735b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:47:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e48074f2db4cdfb7ee8ff6288d59f8\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0b4cd1929b74f9b903471523b85c095 : SUCCESS in 41m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6f0b3c2f82374076a9d4cf6a65fd218e : SUCCESS in 26m 42s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/21e0bdbf5f634275a31ad0a7d814224f : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fca2b95529df4bbba85b5f02acf0cdc7 : SUCCESS in 1h 35m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/57eabd065bd2428fb086ca31a9e62ac2 : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/41bf3cb00a144284aa09ddb432901b8b : SUCCESS in 1h 37m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d4c5de767cdc48459e9c777068a7f2d4 : SUCCESS in 1h 20m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c5964e06379a493b85ec2511e8ecd57b : SUCCESS in 46m 30s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/520ce820697849ab860c140db59e582e : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/754b2f0ef152471d9a034761789bab42 : SUCCESS in 1h 21m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8f71b01d36194bd28be62bf1ad2426c3 : SUCCESS in 1h 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee71571a5bc44aa940a0f97b52e865b : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9112d5d6bbbd4fc8995d3fe5aaa9ea97 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d706d9b427844d0b331898016e8a245 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ac62c1230a438eaf305a85dace42c8 : SUCCESS in 4m 58s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/7fdb26f10cbf4280ad2d6eebd6ba2470 : SUCCESS in 51m 42s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/af3dd799025546e19b4c88595f4f55a3 : SUCCESS in 1h 33m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/965478d4073e4b90b8053fa122d7715b : SUCCESS in 17m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bae00f1479734e398a79e88e24c26d98 : SUCCESS in 40m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9696dbf56e541d8500006f81e53dbef84a9491e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:55:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/044ee6e8cf0c4706abb071d819b57e0c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/38b812a2f3e04685bd92faee3b20585c : SUCCESS in 1h 52m 58s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f06f667726649a58f566b98051be73f : SUCCESS in 44m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5291ff23afc94c0d940eae745131aced : SUCCESS in 1h 20m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/44f49b2ff4c34634921f6ed7cc6940c0 : FAILURE in 1h 46m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/571de8d36c7243b79f38a38290187725 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97527c2442494446b8e5b1c72b9aff11 : SUCCESS in 4m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/73cedfbf13464adb89e6da818937cf78 : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51abff12fc1541aa058e0ed2fab4f8e51eeb11b5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-07 13:26:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7665c7d4ae51877128141646096f6867d3889ce3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 15:19:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d503936e0c1c03a20f9507d233dee8fc037958fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 15:20:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f985b12fd8a84bce63b123f79530b91314b2ed3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:11:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b64a55ae1f6c485dab8dadc7fa169410\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/56876e14cc8648af83c9a718f2dcecc0 : SUCCESS in 34m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d065fed0c7a74477a7716cc2909ed67c : SUCCESS in 1h 06m 57s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b7a52e19d6854b42a1319c649e9dfae8 : SUCCESS in 1h 33m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/22796f68848548c383f32e49ddd9115f : SUCCESS in 27m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/45db9c08f3af41969c3516b80278dd47 : SUCCESS in 1h 05m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/062b5aa88ade4603b968fea2abbe5602 : SUCCESS in 1h 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88a04b78a524b17b781755f3e0245c0 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/543782bffebb42849e56342aaa1a5aac : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5723951dceaa423d9244724cfc95ce52 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bce4c2d37654a3e8178117fd32ac818 : SUCCESS in 6m 29s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1951edf2eec94c129e0c008d7f86097a : SUCCESS in 1h 42m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2ebb83fad6a4983bdf0abdb68b346b0 : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5f188a966a4ab0762593b8e2c669078137ec3e8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c00c31bad6de7217f6cc9d966f5a6f88330b5a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:12:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f1e1621d94453aa78554c42f7f8f63\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f832fc45f640edb8bd5f49bca8111e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59dd65a99ba4614137bae0a6ed5ba719877ba2a2","revisions":{"59dd65a99ba4614137bae0a6ed5ba719877ba2a2":{"kind":"REWORK","_number":1,"created":"2023-03-28 15:58:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/878666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/878666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/878666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/878666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/878666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/878666/1"}}},"commit":{"parents":[{"commit":"9516d00f3b6d6a41bc6782c4a3fc4885f5f1c323","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9516d00f3b6d6a41bc6782c4a3fc4885f5f1c323"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2023-03-28 12:08:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-28 15:58:35.000000000","tz":0},"subject":"Replace invisible unicode chars","message":"Replace invisible unicode chars\n\nThey\u0027re believed to have no special meaning and were put by accident.\n\nTrivialFix\n\nChange-Id: I66ea5711810ed0060cf89192a780bfcc92e79770\n(cherry picked from commit 2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59dd65a99ba4614137bae0a6ed5ba719877ba2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59dd65a99ba4614137bae0a6ed5ba719877ba2a2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
