)]}'
{"id":"openstack%2Fkolla~878667","triplet_id":"openstack%2Fkolla~stable%2Fxena~I66ea5711810ed0060cf89192a780bfcc92e79770","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-06 13:21:00.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I66ea5711810ed0060cf89192a780bfcc92e79770","subject":"Replace invisible unicode chars","status":"MERGED","created":"2023-03-28 15:58:42.000000000","updated":"2023-04-06 14:46:00.000000000","submitted":"2023-04-06 14:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878667","cherry_pick_of_change":878666,"cherry_pick_of_patch_set":1,"meta_rev_id":"b7eb2d23b235f527928ee788177fa4f2bde6bdd9","_number":878667,"virtual_id_number":878667,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 14:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2023-03-28 17:21:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2023-04-06 13:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2023-04-06 13:21:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 15:58:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2023-03-28 17:21:01.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2023-03-28 17:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d5ec7e4215b835949a2ea650d1cdfa62f2b981e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-28 15:58:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"f2b81c57bb0d83f57c350579a52c48b3a4c76c68","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2023-03-28 17:21:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f81d44d8f6172dfe5d0ef6a5beb8aae948212b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:31:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b61a941a6f7f41318dd3403e75162527\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9a428cc0b40c45e5a2fbbf5632f263da : SUCCESS in 35m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/da57d79504f7427b9692aa9bb91b9010 : SUCCESS in 45m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5dc5ee3ed8b249849dde60509cb719ea : SUCCESS in 1h 15m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6a32caefb4f7488bb5af3753ed79deb3 : SUCCESS in 57m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1df6e44737ff4315a261ea76ae8c3fa5 : SUCCESS in 54m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/17db9322c701410fb8bc63c8f145b43b : SUCCESS in 16m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8503d5794c6642b3a217ea05e6fa480e : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffb7dab13cd410299fdb787505683e0 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82d5d2f59a834134ac6912caa67f7a51 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5e8824cac5047e99adf209d549f9e11 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21e5b5174d9f497480c77a9947535622 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b610d887e347749b5572ff23106e17 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f14bd19e2db4e67ae3b0c20d41e5ee6 : SUCCESS in 21m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a35bf73351247118297114dd7d76d13 : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0ba6a1f40114917b03ea1954977ffb1 : SUCCESS in 1h 18m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1e70029dd70446a85449aed685bdff2 : SUCCESS in 46m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12846d31fba34a7a944f80d82e95532c : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3e06620940c0e1cb5865c8696c39154f8eab7903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 17:54:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f86613fa228a483da0ffaa7d79981257\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/91f4a148e2c940509aa9363351daa4b4 : SUCCESS in 44m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9202960325748759a2bff56a94e6c17 : SUCCESS in 1h 50m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/866741ab1900437d9d7e06c25ac76c94 : SUCCESS in 1h 53m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b738ba729c3f40a4accd3db618b9a731 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c08e5ffc8e14b1481b8369e01c7719e : SUCCESS in 4m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f1f1c5c2ca74e20a7583b55bf3db23a : FAILURE in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08728cf710fb286bdca1b74385a385ce7797ed2e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-06 13:21:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"312fccbeae0db94a88fa11d0f927ba4f182c1bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 13:21:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"efcd4bfc56212d215f9ab6b39625e290b104e2c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:44:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3184398522f40d19d4d0347deaeba7a\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b920fe48937347a2bf20ceddfeeb55bd : SUCCESS in 28m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8611234ce4dd44f39e664c0407b3b786 : SUCCESS in 1h 15m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795c38d1d2aa43a38451974158832af8 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/520b404ac4a443d8ab05204acfc4521b : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef88a5c1a228422bac29fd57538060f4 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d217dc8a724e29b5d60317f96c9b7c : SUCCESS in 5m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc20e1c6b31941af9c6ea63df875aadc : SUCCESS in 19m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/03e8da745693439a9d302cffb53d3fc1 : SUCCESS in 51m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cd83753eade476a898b1b51ca4ffb25 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"91d068aa4a49824ab105761909c033c98db9790d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:44:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b7eb2d23b235f527928ee788177fa4f2bde6bdd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:46:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c101c44a010e4a9085fc9fc15f8582dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ed821b9cc44a2c8164d4cdffff5536 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32cbb1431d970a4015e20e2862ca83e156612b39","revisions":{"32cbb1431d970a4015e20e2862ca83e156612b39":{"kind":"REWORK","_number":1,"created":"2023-03-28 15:58:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/878667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/878667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/878667/1"}}},"commit":{"parents":[{"commit":"9a86a7bf071ffa31feb3d0eaf35be31faeb20008","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a86a7bf071ffa31feb3d0eaf35be31faeb20008"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2023-03-28 12:08:42.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-28 15:58:42.000000000","tz":0},"subject":"Replace invisible unicode chars","message":"Replace invisible unicode chars\n\nThey\u0027re believed to have no special meaning and were put by accident.\n\nTrivialFix\n\nChange-Id: I66ea5711810ed0060cf89192a780bfcc92e79770\n(cherry picked from commit 2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32cbb1431d970a4015e20e2862ca83e156612b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32cbb1431d970a4015e20e2862ca83e156612b39"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
