)]}'
{"id":"openstack%2Fkolla~878767","triplet_id":"openstack%2Fkolla~master~I66ea5711810ed0060cf89192a780bfcc92e79770","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2023-03-28 15:50:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66ea5711810ed0060cf89192a780bfcc92e79770","subject":"Replace invisible unicode chars","status":"MERGED","created":"2023-03-28 12:11:31.000000000","updated":"2023-03-28 15:51:47.000000000","submitted":"2023-03-28 15:50:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878767","meta_rev_id":"ad9789b8735c238d30612581411f3854f3d77f32","_number":878767,"virtual_id_number":878767,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-28 15:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-28 13:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-28 14:31:24.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-03-28 14:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-03-28 14:04:11.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 13:41:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-28 13:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 13:57:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-28 14:04:11.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-03-28 14:31:24.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"465a5dad5e1bf468e193e6b530902a151e1685fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2023-03-28 12:11:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a0b102b39604ec712cb553f0ee53f8f099ac0ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 13:41:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd92f1950b5946f4a0a65fa4effeb387\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/40b8460e2b764f58a82e734889df9dc3 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/909cb02f622c4b53838e05142491edd9 : SUCCESS in 20m 23s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d353ed42bab441ea6d831fde1b63b12 : SUCCESS in 1h 26m 40s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35b948bcd3c04e23a3c7de94259133d8 : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06be79f2ba7f4f8382b58cad19b53088 : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/834c048a4d954d94a5c6bea8b9906936 : SUCCESS in 17m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9fff34637cec4dca90abceff626b2ca8 : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30749bb834723f05a451dc2e8c1ad617797fd388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 13:47:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7af821f932004a12bcbb83411844acf5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/20d1474c6a1f4171be91fa165380343f : SUCCESS in 23m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a8ae766253054734bb33be6c815509b0 : SUCCESS in 1h 06m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b0fba583d2c44f3ac40343a064b7034 : SUCCESS in 19m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/36f91c141ca34c6390087d9b02a19352 : SUCCESS in 55m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bc66b46517bc4a93bf5e72368925e437 : SUCCESS in 1h 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b663839fa7400584a2bfae2ca5972e : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8f702d694b2474c8f8400240aaec932 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84a22857758744e1a707593a0a722fb9 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a6229bd88c3401097b511fb9684ce8a : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8855153b7b4a77b4b9ed4f210696de : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f95b9318589b45ca8c4f0ac0efe02351 : SUCCESS in 15m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/96ac7fc0305f4f1bad1b35135c62c890 : SUCCESS in 1h 07m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e5a3851cf32844a5aef399d9f6a22a16 : SUCCESS in 1h 28m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/791fa6bacf124931942b9f28a59072bb : SUCCESS in 14m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8297a4ff91904eadbc2378065f3db68b : SUCCESS in 55m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/13c5d05d773b44e4842e5bad34803949 : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d24a416760a15b32c760614433d08615757d5fc4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-28 13:57:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0e009a086d8479196eb2bd2fb9447be12b131aa","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-28 14:04:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdaae92891ae16b1fbabb0621ecf0a3a9bf35473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:04:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f453a5e028e86c23db34f793841097a53df7cef","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-03-28 14:31:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3c2cf386c56d4ea38e2885d19a7b6f627650742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:50:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/873c0f8d23f04b3994be90f4bfb5ae75\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f13997036e8e44d28a0f8d0399d52f25 : SUCCESS in 43m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5e4c12a388ab4fd79ab3a7a1712cef30 : SUCCESS in 1h 10m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c5df30aeaf6e4f019009c5e247b5d4f1 : SUCCESS in 1h 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15e38e882f7429a9c08c3ce996655d8 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3398ca94160e4c7aaa623d7fa097d715 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/378922a527f94a2891fbdb2eaadd4e16 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afeda2fe15c64e7083f7bbe06a0339eb : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fd7e90ca3c9c4982a404f024f1c17522 : SUCCESS in 16m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4806c2d71a024b08aad7b72b05fe2dbd : SUCCESS in 1h 10m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a45acd0216284ad2a748da3afeef69cd : SUCCESS in 1h 40m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40a2fccade9e40d3a930fc1bf24c2928 : SUCCESS in 44m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c6675c976e654694b7fa2818615e5ada : SUCCESS in 51m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0781303b118f4d3a8598197551bb740f : SUCCESS in 1h 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cc8ccf959be932b6631d80e7e8de4326873e26c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:50:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad9789b8735c238d30612581411f3854f3d77f32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 15:51:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab0725d3f114f47bdd42c7130102969\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787c2be56cdc447d97a100e7082d6391 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421","revisions":{"2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421":{"kind":"REWORK","_number":1,"created":"2023-03-28 12:11:31.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/878767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/878767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/878767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/878767/1"}}},"commit":{"parents":[{"commit":"41c489715fd676384f766547252b9895fc593d91","subject":"docs: Rename Community maintained to Untested","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41c489715fd676384f766547252b9895fc593d91"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2023-03-28 12:08:42.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2023-03-28 12:09:01.000000000","tz":120},"subject":"Replace invisible unicode chars","message":"Replace invisible unicode chars\n\nThey\u0027re believed to have no special meaning and were put by accident.\n\nTrivialFix\n\nChange-Id: I66ea5711810ed0060cf89192a780bfcc92e79770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ddf0aeb204d754636ded6ff2be4e2dcfc3ee421"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
