)]}'
{"id":"openstack%2Fkolla~878822","triplet_id":"openstack%2Fkolla~master~I989cedce7a9af3882b5510d3479b29150139e861","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-04-06 10:33:36.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-04-07 11:14:50.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-04-07 09:48:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I989cedce7a9af3882b5510d3479b29150139e861","subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","status":"MERGED","created":"2023-03-29 09:19:53.000000000","updated":"2023-04-07 11:15:48.000000000","submitted":"2023-04-07 11:14:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878822","meta_rev_id":"c7258c66706482638b8a6cf38b773b5f58a19150","_number":878822,"virtual_id_number":878822,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-07 11:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-04-06 13:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-04-06 13:14:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-04-07 09:48:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 10:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 10:51:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 09:52:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-31 11:39:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-03-31 11:39:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"47a5e95e8bba73456e4142a9860e4ed7a1451054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-03-29 09:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23f1413eb87dacd04a02d9691616c92db6adfaa5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:48:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39d8472f640047118b1aef3defb4887d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a5979474ec084e19b3556279d2c2be08 : SUCCESS in 18m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aa1ab5d0ddb24f6a8249813d99e7e358 : SUCCESS in 16m 12s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6eecd530d8ae4fbc80fe525331236726 : SUCCESS in 1h 25m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b9dc75ccce94193be260fe86d0ea63b : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9adf5e5e0cdb4996a13d8a5a9d6b31fd : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0d04522c901e47b68330f5d2819a62a6 : SUCCESS in 16m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/091d60eae1224fd3a8240a086f8616ce : FAILURE in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccdcc2879109de8f1e3100c10aa7a2f09b4a36a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:51:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a87568475d4b5397753836662a6010\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fa52c6e2673943f2b087995db058a3d3 : SUCCESS in 19m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/065bd220cc994faca0ad1827166a9171 : SUCCESS in 1h 20m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3932f776706949b9a0817a7759405aa8 : SUCCESS in 17m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/82d850448d874094ba57b3d44d45d8dd : SUCCESS in 1h 09m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d601b34602e14a8ca7ba8472610efaba : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f51d134754b447a83e3b20dad9a0f0c : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/897368b9bf8144a7adaa0fa7d36226fc : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2af8b51650c04ba2b3d409d3ee27eadd : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7deac9b94a224247ab7f5c8969103acd : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9daa460fe6f54c1c8f6e9516d889b20c : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7cd850f23e05466086cb1d2d4c2c652f : SUCCESS in 19m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e895a704f4b43f3838ed403f6211ced : SUCCESS in 1h 14m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c6403bf4f1fc491dbc7fdf8748ba45f3 : SUCCESS in 1h 25m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ebfd0ab8473044d6a5397b391a8e015e : SUCCESS in 14m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0cd172be91014092b0ff0c3ec4387939 : SUCCESS in 56m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9ef4f1e3806f402594eab243180fb2fb : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"609add08fea08ae002bd2bdb731a40d8dc0630f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-30 09:52:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c4079c0a43c6858a30c99814deb5940bb1a78dd","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 11:39:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66f31f8222a7486f9860cb1acd0a117ff2156af9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 11:39:22.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"084f0dee25e77134937d734d949104ed72c3b3ab","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 11:39:30.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"f6d5d664816031c2ba23803ead2702a4377f70e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:39:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dfb71bfb77560732f03623f526124affb3bd9c4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-31 11:51:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b2634146a2a03a5238dbfc916cfbd61cd889347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 13:00:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6af1a8f6f4743cf9e40b3ec4cbf8413\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/06169e35b7af4fed8808f412c4d189bf : SUCCESS in 18m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/695626fc17414154a17c602be711f23b : SUCCESS in 58m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/66186623dd274c10ba56c7e7e5c5470c : SUCCESS in 1h 07m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1dbf71b51948fdabdfee29637efdf4 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6dafd9974f54f659d7416dab534b480 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a74100b1d384fc39cebc12cbf9dcf1b : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0063b8bba774d6b8be257f8370bb72d : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2101f055aecc4e0fbdf4e6d698c29fda : SUCCESS in 19m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/32f58a9efc9b4e079a8d6de1648dd7d6 : SUCCESS in 1h 12m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d929c2845d2b4a6c826e93b41af6077d : SUCCESS in 1h 21m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3f9ea41e2d53450dba97c791682840fe : SUCCESS in 31m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/424f0a162b914a5aa5152f86759c72a0 : SUCCESS in 56m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8cd32b5be1e048f28877d596b01af0cd : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"280ec638d39b0a0d849b65ea5211daea9bf5293e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-04-03 09:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f64451ff5b1e8d4e8a8802d1ce175db493b6b53d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-06 10:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8b8182f045ac282d8d67b2560b8bc848be55a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 10:33:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f96347f79ec29ee14aebfd04f3efb424f382dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 13:12:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82fa508fc42c49f7a0fd751e8c7824f5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e015e69ff0574d9197f6de4b4f30afe5 : SUCCESS in 14m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3e9a7b33eab94e51a03e51a6c9c77fe1 : SUCCESS in 55m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9be80b3d5554531811163f669cc083a : TIMED_OUT in 2h 32m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805580c898914716ad0630fa344540a3 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f798ea7feec414c866b4a97ceba092f : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3831b959ce5d477ca668f90211e355dc : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059e9d3d290e43ffa8b3c76dc480e066 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9b17dcec64e040119ac99918612948ea : SUCCESS in 18m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/eb3e32073e19431b8216564a82dcdef0 : SUCCESS in 1h 10m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7a66c873b0a94baf8c927fcc6d8f7795 : TIMED_OUT in 2h 32m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1053cc26d2814c3f90a8d16e8770db01 : SUCCESS in 18m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cb68812e161d4cabb74815f88e9c59c4 : SUCCESS in 55m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3bd39ca684204ef5908475c9dd50ca1f : TIMED_OUT in 2h 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6a3439c80cabe5161c780c29f13d0b8c0314132f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-06 13:14:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f0c241fdfaf25c0b889484187333731eb146608d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:34:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9dd0809412498a98a5fa17e9975a2e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bd8614a14ca0409aab00a826ead56f04 : FAILURE in 17m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb142f278cf345d3bc53365a57411811 : SUCCESS in 15m 53s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b26bcea4bff246b4b28a339df3010833 : SUCCESS in 1h 18m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66d8fd93014d4bffab43654d40b51626 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d747a63c7ef74071a8912879d7473a91 : SUCCESS in 2m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7a0a378d49974de28ad1a8749a433444 : SUCCESS in 55m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eed6445f76f84d8193c6a5138c45746d : SUCCESS in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1816127d33eb666873ad1bfa6248c6c61a9e798f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 15:51:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd616ad8967c49bc9e79ed349d49674b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/70aef9bcf178465995d79dd1ae0ae9e2 : FAILURE in 16m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dfc4dd5e45be4101a3e9e507d1400a33 : FAILURE in 21m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6fc385f06cdd410486a6c9ff6fa254fc : SUCCESS in 20m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/18e27e6e721941939fbbe575b0237800 : SUCCESS in 1h 12m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bd2bb3f0cc484da3a3445106c2e141b0 : POST_FAILURE in 1h 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be3bd324cc994ecc93ae523d83eb42c0 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/497b725173b24b9c9b816eb74858daf5 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c402ba5b7aea49028e16b4bc8ba6adf2 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c9ea7ad5167477eb4031792607c92e5 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde80c3d02c24dadae613b637070b7bf : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dd549890ab084c0d8703350656a7c051 : SUCCESS in 18m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4ffe69d5da7a4fae9b94c7365856125c : SUCCESS in 1h 02m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9e5df41fa0ad450cb8aea235b09b71ab : TIMED_OUT in 2h 32m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fe77c0c5c61c43e1a2a7200463ee427b : SUCCESS in 27m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1a67cb0a753d4caf92c1858ecab27562 : SUCCESS in 55m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/db2055b598d6425799342fb6c2c2ae2a : TIMED_OUT in 2h 32m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b22e50abaf61233dc703e76f8e94dcdfe9e5a53d","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-04-07 08:07:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dcf34152eed0094ad173bed879e8d481a10bd0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 09:44:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de92ce5631db45f2b3aef4fb551ab195\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1af3c30a8cf84f82a1fc349609953c61 : FAILURE in 16m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ba77d6d7cb2b4590b634c54a3e6c666e : FAILURE in 21m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3cfa8db45107401c9d52ec6639b9b6aa : SUCCESS in 21m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bed7fe926dd7493284fd6e8cdb3f5f19 : SUCCESS in 46m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/55fab6171d2845e7bab35fdd7be02712 : SUCCESS in 1h 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42bbf6fb8e4045758eb76e2e4b989512 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42d05ee783d9493e913472bb5fc2b210 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/558775c2320f4470b3c99908dade2db1 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ac0a90191b4b5f8428a1e3e7de951f : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0f14185de14dbd971c55450de5c84c : SUCCESS in 6m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/728da1e66d9e4810bf4c55981c245a02 : SUCCESS in 20m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0acf68ad31af4a619be4bb6e326cc9a4 : SUCCESS in 1h 10m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fe0332a981bf4ace92f473b05566474e : SUCCESS in 1h 21m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cacbb3da35e7438bba6d7b49f066e335 : SUCCESS in 16m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2431e31177e24f84b9b0e6602c720276 : SUCCESS in 53m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/18a1daa9249d4b94ba886947fa447ea4 : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2ed27804199449446f066ebf4743ffb4d3e088ca","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-07 09:48:59.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3fc0d4367a1923ec51947220649809a437f22890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 09:49:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd6545f739e66f1faac7e003c30826c67f00bb64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 10:04:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b01f15ddc94b8390c5ef2746476bf0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/98d9b77f43b241fba4674b2db5bb4dfc : FAILURE in 22m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7faaff62ef3047c083b2b4783b0f0599 : SUCCESS in 22m 21s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6108139059d849dcb25d13ef6dd95033 : SUCCESS in 1h 54m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/621f0f0a709242f982fe8a8eea8a20a8 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6870a49bbeb94ee1ae38619b21cc593c : SUCCESS in 2m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b6007935482549ca862fac1b15b6cfc6 : SUCCESS in 25m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/655e1edc098f4d969de1acbc3fdfa3d0 : SUCCESS in 1h 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"579c93a98e34e536c875f8075e7a830928e7672a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:14:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0313f2fb5f364082928c8b15bcf87081\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a7614208c84d4e9a922634cad0b7897e : SUCCESS in 21m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dbee2e6a2b2142d6a087ef22d20144ce : SUCCESS in 1h 10m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/23a05cdb9af242fb9eba186411577f30 : SUCCESS in 1h 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c5b300baed422bbff31032ea3c97bb : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f15e3043e394c44ab4e051e922acbc9 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f8c290854cf46218caa8e3b1c045ccf : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434a0c34c21547d280a0e6cd0f15d387 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2203031afcdd4168b2a0db5751af0c98 : SUCCESS in 19m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9393a38e99643f3bcb16ed45d439f4a : SUCCESS in 1h 10m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9b1cd67e45334022abfcdf4a19a16036 : SUCCESS in 1h 21m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/12a76be766e24e9da225a3a03255330c : SUCCESS in 17m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/20791e912cc04eecb05342e87ca53935 : SUCCESS in 41m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/26d9abeb5ce4445fa7bc254881630f4c : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9b6bbb99a2919032c3077fda47c5fd9cdecca88d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:14:50.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"c7258c66706482638b8a6cf38b773b5f58a19150","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:15:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2968175788df4f04870c1a1a7bcfcd67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bd91e2945d4bfb90cbc21727feaec8 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2783fb710710103fe6aca3ff18c976590ca15e09","revisions":{"2f59106b6bd1cc6fea75358b9c94180fff29e7fc":{"kind":"REWORK","_number":1,"created":"2023-03-29 09:19:53.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/22/878822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/878822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/878822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/878822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/878822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/878822/1"}}},"commit":{"parents":[{"commit":"e559f2194b847efc65a9ec0339667764a48d1e87","subject":"Merge \"ovsdpdk: add libdpdk-dev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e559f2194b847efc65a9ec0339667764a48d1e87"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:15:32.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:19:51.000000000","tz":120},"subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","message":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\n\nCloses-Bug: #2012821\nChange-Id: I989cedce7a9af3882b5510d3479b29150139e861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f59106b6bd1cc6fea75358b9c94180fff29e7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f59106b6bd1cc6fea75358b9c94180fff29e7fc"}]},"branch":"refs/heads/master"},"2783fb710710103fe6aca3ff18c976590ca15e09":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-06 13:14:06.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/22/878822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/878822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/878822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/878822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/878822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/878822/2"}}},"commit":{"parents":[{"commit":"32922bbbaedd6aa257225674acc6252905cb5af0","subject":"Merge \"Revert \"rabbitmq: Pin to 3.11.11\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32922bbbaedd6aa257225674acc6252905cb5af0"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:15:32.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-06 13:14:06.000000000","tz":0},"subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","message":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\n\nCloses-Bug: #2012821\nChange-Id: I989cedce7a9af3882b5510d3479b29150139e861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2783fb710710103fe6aca3ff18c976590ca15e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2783fb710710103fe6aca3ff18c976590ca15e09"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
