)]}'
{"id":"openstack%2Fkolla~878825","triplet_id":"openstack%2Fkolla~master~I01578289e65010748bfaf6d6b81861764aa62368","project":"openstack/kolla","branch":"master","topic":"keystone_fix","attention_set":{},"removed_from_attention_set":{"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-03-30 11:17:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I01578289e65010748bfaf6d6b81861764aa62368","subject":"keystone: Fix for missing libldap-common package for debian","status":"MERGED","created":"2023-03-29 09:45:43.000000000","updated":"2023-03-30 11:18:19.000000000","submitted":"2023-03-30 11:17:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878825-keystone_fix","meta_rev_id":"75883e8142064d5f572f9569bfcc743a0dd30b17","_number":878825,"virtual_id_number":878825,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-30 11:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-03-29 13:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-03-30 09:51:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-03-30 09:51:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 11:25:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-29 13:23:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-03-30 09:51:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5cd894db23052d2926ebbb55e860a2bc20bc7db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-03-29 09:45:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5e3e4a51767eacefdb0261c91ee142271fd34eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:25:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28853fc25e6044d7b80de8cce58ca4e1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/45d9d05da3514f0888c0ea85c83d109c : SUCCESS in 27m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8a0eb323811b4b178df2f853ddc39a90 : SUCCESS in 1h 22m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fae2dd03067a47399b671fd5a9f4549d : SUCCESS in 20m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/390ad3c5f8f445e282f3b8b98c237e35 : SUCCESS in 54m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e2d29cb488ee43c4bfb5496ba58da79d : SUCCESS in 1h 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a821ee7e7ae047959515c635051b265c : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27ba1da7c1d34372881b47a4de25f87a : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40790ab7623a4461979cf12e71e144b9 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63032b847e6d451e96d3fdb1b8038a4c : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aed2a166c954a6aaf3e8781fd14be95 : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9baf1f86d35247ca80c1ecd5600c43f9 : SUCCESS in 22m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8e5ee96a3a6f4dc1b25ef774f06da887 : SUCCESS in 1h 15m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/97da368e59ea44bb8f88855a84d7560f : SUCCESS in 1h 24m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9014a8a51feb4c14adbb487e86cd814f : SUCCESS in 18m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c60be7f083d7436faba041e00d2483a8 : SUCCESS in 50m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7eedf8bc1d0840b39ac4dfb88f77c587 : SUCCESS in 1h 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"830f67ad6de937c0e43c5f29ab47d9923da734fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 12:49:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d254a08a9e294dd2a766c757cb2db81e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8885073eb18f41c2975c9cf55fb1e172 : SUCCESS in 1h 15m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b93fee8c01414ec5ba780c6a59789dfa : SUCCESS in 1h 08m 47s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c9238656b6b142748def925a935fd52e : SUCCESS in 3h 00m 28s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cb4dda1d4ee4ad8882e4a10dc8fb0b0 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c6f557af7fa48b08dd4e60bced464bf : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8e41ca3212a34b24891f16245db8dea2 : SUCCESS in 16m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bd3101bacbb5463ead00c621652a03b5 : FAILURE in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79742e23c02905fa37485310c08d9ffbc17ed8cb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-29 13:23:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62e119dc0171a5effd1e291e34fc9a6585359e8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-30 09:51:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c0495bd10b625ab7026339c6ce9fc7d6d819306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 09:51:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d34d32677ceac4ab51732b6e6c744b05d4d8e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 11:17:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3a8391bcdcf4e8e830708d8765e2847\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/de1a676358bf4e8b9cef01f012598536 : SUCCESS in 20m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/226c9b3e36804c83a4e47f4aba9db0ff : SUCCESS in 1h 01m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a3ef02b48a8461491b9cc3a99845fd1 : SUCCESS in 1h 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f67be6214ff0451ab15b9911f521d223 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dcff0adfdca428facc205fa019b5342 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16901233778e4f8aa3a70e7492efffc7 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc78837bf1a749dabcf4b30b9c9d3609 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/564c15873203445abd9de1e381f908fd : SUCCESS in 15m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ea77183c35fa44a5b893560cae551b2c : SUCCESS in 1h 08m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7f51fa4dba354195a1948c914e0f7c00 : SUCCESS in 1h 20m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f0847c493d09402f8fe8a343b925d444 : SUCCESS in 20m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/63b2ad2f53a94ba7a0bf30d5e1783d11 : SUCCESS in 51m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/32d35d9369cf4fa189a0b937a6318095 : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"281a98cc2f1d94245bb1b9987f6473f792781394","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 11:17:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"75883e8142064d5f572f9569bfcc743a0dd30b17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 11:18:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc44174c9ed04af79593a640a2236d50\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedd00cca7f84be9a014bb3fc895c392 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2050aa35dbd074610adee5611194fde280cfea20","revisions":{"2050aa35dbd074610adee5611194fde280cfea20":{"kind":"REWORK","_number":1,"created":"2023-03-29 09:45:43.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/25/878825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/878825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/878825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/878825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/878825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/878825/1"}}},"commit":{"parents":[{"commit":"e559f2194b847efc65a9ec0339667764a48d1e87","subject":"Merge \"ovsdpdk: add libdpdk-dev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e559f2194b847efc65a9ec0339667764a48d1e87"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:44:58.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:44:58.000000000","tz":120},"subject":"keystone: Fix for missing libldap-common package for debian","message":"keystone: Fix for missing libldap-common package for debian\n\nCloses-Bug: #2009484\nChange-Id: I01578289e65010748bfaf6d6b81861764aa62368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2050aa35dbd074610adee5611194fde280cfea20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2050aa35dbd074610adee5611194fde280cfea20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
