)]}'
{"id":"openstack%2Fkolla~878841","triplet_id":"openstack%2Fkolla~stable%2Fxena~I88e1e0ca01b2b2effad9515b905cc761fbb5d2d4","project":"openstack/kolla","branch":"stable/xena","topic":"bug/2013075","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-04-06 14:44:03.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-06 13:20:44.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-04-06 14:44:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88e1e0ca01b2b2effad9515b905cc761fbb5d2d4","subject":"Support qemu block rbd in nova-compute","status":"MERGED","created":"2023-03-30 10:55:15.000000000","updated":"2023-04-06 14:45:12.000000000","submitted":"2023-04-06 14:44:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878841-bug/2013075","cherry_pick_of_change":860230,"cherry_pick_of_patch_set":3,"meta_rev_id":"b69c8973e97d118adb34e45a9ddb95d5fa581529","_number":878841,"virtual_id_number":878841,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 14:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-06 13:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-03-30 18:41:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-03-30 17:44:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-06 13:20:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 10:55:15.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2023-03-30 12:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 17:44:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2023-03-30 17:44:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-03-30 17:44:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-30 18:41:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"ba4447305c45e12c563044a539af809705f4a520","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-03-30 10:55:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/nova/nova-compute/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"9d10f9ec6bd6c728f10dc7e0618b5bfae02aa4a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-03-30 11:04:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"553246c85f0252f89b8e1d6cadadd1d6c2a65cd2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-03-30 11:12:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c4b20d797b14dfd4c3a84df9a5b2482f60d72061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:19:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a244201e7106435bb720a070ad21200e\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d7d6e880d2054aefb56dc40d9cb3a74f : SUCCESS in 36m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cfaf7346d5ad4e66902b409c94d9d9f5 : SUCCESS in 33m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c9053da6afe54de49a0af57023e2b109 : SUCCESS in 1h 08m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cd4fc5582deb4cd6b0c53974c744ffec : SUCCESS in 1h 04m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/57e777f63202428f9c41909cf184ec50 : SUCCESS in 19m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/704dc78dbb1d4e9cbcaf63b02089ade4 : SUCCESS in 20m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/73919dc3dbdc4d78b7d562af58cbddb5 : SUCCESS in 1h 01m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca949d4eb8c4bef91dbe880757627fc : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf43abd5c9e4041a0ccabf176ee7585 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810b7e990f5d4ec78c9f4691f82ff68e : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac97e0bd282d45d4be8a81706e6a5c16 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4766ee548f0e45e6947c1b5876ec61c9 : SUCCESS in 4m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/926ffb2d88da4735bcbccc27d253924c : SUCCESS in 22m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a58f0f74a794f23ba2acc4e6618fe70 : SUCCESS in 22m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2862b17c1ea436d91601364cf7bff61 : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/17297cace4d04f6292dd28bd4df4b4a4 : SUCCESS in 37m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/10ebdb02c0ca4ba381b96e0e1065190c : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8cae6a149be51f7d4722ec499e4714c15ba5f65d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:56:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d875ca4c4bf4ac6adf00be51291ff5e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f2af8e2181a34456b662a6b7bf5f0529 : SUCCESS in 23m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4573e6c725444adb9d5180660ffc640e : SUCCESS in 16m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3330f287fbef45a29d1ac432ce0a40e3 : SUCCESS in 2h 48m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83f1ed7ba2d1403997591b22bdf00111 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fde4ec26b6494cc38d9e319c78bf01ef : SUCCESS in 5m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b88315737b6749cc89126590db49bfe4 : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d456c50b9939b4160fff4b0cee6944071b25fb7d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-30 17:44:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4d65456093e634e539cf1ce58103acafa74dc2d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-30 18:41:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"64b18f270c7a6fb5db845de10ed93d5aa2fedd1c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-06 13:20:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42a485583b7845987d82022217a7419ec7b24466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 13:21:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"447ce91ddd9f08305adca0f75776fe94cb8cf21c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:44:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab5b8eac3d24af4afc9abe93ebe562c\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/503a395529b14d669b6917a2176b7f7a : SUCCESS in 31m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2efa45b7138347d1973905d8989d3b86 : SUCCESS in 1h 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8e95a28d6d405a99c0df013f3c9324 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9023f88c18246b9ab86ff8ed459339e : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b91efb52c8c44cf89a5fc65e2831ccc : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f3758b361542228258960213334e56 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c62f8fee8b384ef1aabefe8a85b892b4 : SUCCESS in 20m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/772c4fe50ba14569839a4187ae5e11cb : SUCCESS in 55m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fd5af9dd0d74500a711e71bc478badd : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":2},{"id":"af5bc7723ae8f90b53079d5dd794fac812e47130","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:44:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b69c8973e97d118adb34e45a9ddb95d5fa581529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:45:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e59a3f655e6b448bb405c8b3064fe171\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f41e589c6314ba987bca1eaaa0c122c : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ffa0d5251b5c1623cea23d3c8f72b88f40dabf56","revisions":{"df8d16879927980a9a111204f0a9ae12ff94da04":{"kind":"REWORK","_number":1,"created":"2023-03-30 10:55:15.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/41/878841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/878841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/878841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/878841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/878841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/878841/1"}}},"commit":{"parents":[{"commit":"9a86a7bf071ffa31feb3d0eaf35be31faeb20008","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a86a7bf071ffa31feb3d0eaf35be31faeb20008"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-04 10:07:40.000000000","tz":0},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2023-03-30 10:55:15.000000000","tz":0},"subject":"Support qemu block rbd in nova-compute","message":"Support qemu block rbd in nova-compute\n\nThis is necessary if using nova-compute with RBD backend (directly,\nnot via Cinder) and wishing to make an instance image to Glance\nwhile (1) following the recommended Ceph permissions which do not\ngive Nova write permissions to the images\u0027 pool or (2) not using\nRBD with Glance or (3) not revealing what Glance actually uses\nas the backend (hiding locations).\n\nCloses-Bug: #2013075\n\nChange-Id: I88e1e0ca01b2b2effad9515b905cc761fbb5d2d4\n(cherry picked from commit 9e92e4c8e494ec345ccce1eff739f18bf9035c9b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df8d16879927980a9a111204f0a9ae12ff94da04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df8d16879927980a9a111204f0a9ae12ff94da04"}]},"branch":"refs/heads/stable/xena"},"ffa0d5251b5c1623cea23d3c8f72b88f40dabf56":{"kind":"REWORK","_number":2,"created":"2023-03-30 11:04:45.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/41/878841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/878841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/878841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/878841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/878841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/878841/2"}}},"commit":{"parents":[{"commit":"9a86a7bf071ffa31feb3d0eaf35be31faeb20008","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a86a7bf071ffa31feb3d0eaf35be31faeb20008"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-04 10:07:40.000000000","tz":0},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-03-30 11:03:47.000000000","tz":60},"subject":"Support qemu block rbd in nova-compute","message":"Support qemu block rbd in nova-compute\n\nThis is necessary if using nova-compute with RBD backend (directly,\nnot via Cinder) and wishing to make an instance image to Glance\nwhile (1) following the recommended Ceph permissions which do not\ngive Nova write permissions to the images\u0027 pool or (2) not using\nRBD with Glance or (3) not revealing what Glance actually uses\nas the backend (hiding locations).\n\nCloses-Bug: #2013075\n\nChange-Id: I88e1e0ca01b2b2effad9515b905cc761fbb5d2d4\n(cherry picked from commit 9e92e4c8e494ec345ccce1eff739f18bf9035c9b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffa0d5251b5c1623cea23d3c8f72b88f40dabf56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffa0d5251b5c1623cea23d3c8f72b88f40dabf56"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
