)]}'
{"id":"openstack%2Fkolla~879177","triplet_id":"openstack%2Fkolla~stable%2Fyoga~If0f9b4dd058a257d0653332d1b663e150c717304","project":"openstack/kolla","branch":"stable/yoga","topic":"test-malicious-tarball-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-03 14:55:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-03 16:22:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0f9b4dd058a257d0653332d1b663e150c717304","subject":"Fix test malicious tarball fail","status":"MERGED","created":"2023-03-31 19:11:00.000000000","updated":"2023-04-03 16:23:19.000000000","submitted":"2023-04-03 16:22:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879177-test-malicious-tarball-stable/zed-stable/yoga","cherry_pick_of_change":879176,"cherry_pick_of_patch_set":1,"meta_rev_id":"1c463359da47cac12d707fe9b8313caabe3b6bb3","_number":879177,"virtual_id_number":879177,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 16:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-03 14:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-03 14:55:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 20:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-01 11:54:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-03 12:14:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"3eab27b25a985e34e1dac3b0b8948103a2210709","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-31 19:11:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py","accounts_in_message":[],"_revision_number":1},{"id":"8311d31acb2abd116b7adcbdec2ca9a8ec32d59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 20:45:22.000000000","message":"Patch Set 1: Verified-1\n\n(23 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a18c30fdf0334ce9b57f336b42e953dc\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/49d1158ff0fe4d72af2d3dc6460349f9 : SUCCESS in 28m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/36cb58315e8e409eb255848a3407a5d6 : SUCCESS in 37m 51s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/68aea3cc51d144539f8c083109a51093 : SUCCESS in 31m 04s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/638c40c6d92344c9afa66fda43c330c5 : SUCCESS in 1h 09m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8f913646cabe483ebb6eb763112e8053 : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/65afd4c3177741afa897d4575fbef065 : SUCCESS in 1h 11m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/786ea25b32a54b00be58715cd18cf000 : SUCCESS in 1h 19m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ae3b3d45a7a406395dd7138e7f441dd : SUCCESS in 16m 59s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5dc930555c5f4f0784dad2fb387e25c8 : SUCCESS in 16m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8263ade7f33f4e4aad0dc55b4289dd29 : SUCCESS in 51m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/19a2f0f70013429aa41645e4932e19a4 : SUCCESS in 1h 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbdb0916227344ec80fafd5147b1ad38 : FAILURE in 2m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f11c04fe6e3142109e2bc8a73077e5cb : FAILURE in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67c50dff3c84f4ca403f9dbc0adcdb8 : FAILURE in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639e87560a454ac3a6a6ed932b524049 : SUCCESS in 4m 49s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/cbfcf9b96e6b4e58b083226faefabe99 : SUCCESS in 20m 20s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b4da5b0819e64d75b3e2fda10758ac53 : SUCCESS in 1h 11m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5636ddd8e2dd48a796d5683ddb92db9d : SUCCESS in 27m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac49d3a730fa428a897a5357d4e7437f : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4c4d2f1834feb2389ee464e0c472383c8d616ea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 21:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32b34a9688c34a4391779579b6537f53\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0779215bf302499b9e501f8b0ef3e844 : SUCCESS in 31m 00s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b803503bcd3d430895decc5d7d67081c : SUCCESS in 1h 38m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d929a5b6c0c94ba1bc40b7d4a860ad1c : SUCCESS in 25m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f91a9e02c4b4d1e924ca8e60a677855 : FAILURE in 1h 59m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76a6dcbc88a7442eb20f762349f48f48 : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47302a098a5f4b4db683435fe5806157 : FAILURE in 5m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/93145479a60347d884cc193349efbd95 : SUCCESS in 24m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05c8eb2a22ab20c19e3ba3c3defd08ef443442d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 11:40:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4b24f0607bde1bf0bcec04bc543c937f00b33ed3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 11:40:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1f9eb0f674878f40ece1390717bf66ecf2e0114d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 11:40:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d4830c96f4155641d1a80302c664faf099aa173a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-03 11:42:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e0b05b579dc101318adffc3d4d1b386c726f718c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-03 12:14:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d7140d8350dc8113ad8d288dd3f8913eed5e8072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-03 12:58:00.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e64e7a0ac1a32603b9c231a48c70d920b31e7320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 14:24:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05603e3e339b464cb806a6938b2aac9f\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f1064ed80ede4d71afab33124c41d922 : SUCCESS in 28m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/af1ba9bfcdcd474a844f4a69467a57b6 : SUCCESS in 40m 27s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/3bb1e507b05a44c88c7ce19d4cfd12a8 : SUCCESS in 34m 15s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e797fc6cb6954c1cb292b9cf978f82c5 : SUCCESS in 1h 19m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/836626afcc8b4d9ebd3cdab13a6e7d3f : SUCCESS in 1h 14m 11s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/84c3669351284e37b08e666dfda5168e : SUCCESS in 1h 14m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0aca6fd2908f4d369a5f655b0d6a7328 : SUCCESS in 1h 17m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2ff985b89af4d918c8cd2af67abd020 : SUCCESS in 24m 09s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c4b9f523f5b4ad38c1265af8e65620e : SUCCESS in 19m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/090846d9a03e42ea85af1fdcf7c6e85a : SUCCESS in 1h 02m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/adbcc57781b746308c39d2f319a74bd2 : SUCCESS in 1h 08m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372e002895cf4a6ba8cd9355eb748be1 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f1aef723b342609ddb462e9fc3395b : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc4b4266276f4e60b301b6300b8262e4 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a29596dc2054161bacc2c9b1782b389 : SUCCESS in 4m 40s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/9499e8851a47462abccd395a4e1e441b : SUCCESS in 19m 52s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f97628fa8bcc4f319e22c043f36ba468 : SUCCESS in 1h 08m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/58935f62326e4a85a58c35afbca508eb : SUCCESS in 21m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6555690392eb413b9b9a81ad4bdb1efe : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":3},{"id":"acc6c15794930402fbe2141242e1a5d8de51a1af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 14:55:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"34eb4ab44973d54967cf9ce6eef2059a4a4b3fab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 14:56:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f4f10ee33ecd2a2e4432a27905e9faa9fefaea0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 14:56:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1083eb83664d51ab248d30eb5fb74f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a7ee7f216ad4917a0dd39acad1c03f3 : SUCCESS in 1h 56m 24s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/22667e791c294c1ba7717afb732b0bd4 : SUCCESS in 49m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf68bffa86e344a99aa6d8036e6ed394 : SUCCESS in 1h 26m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e47448bbfe34c739657370264dabdae : SUCCESS in 1h 32m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70607113be2a4deabc9dd90eff28d783 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb189d92d0364377bb16f9ee507fa643 : SUCCESS in 3m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4018b26f83b842ffa64e9dd76b8af850 : SUCCESS in 1h 07m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c293bf64f47398e9b75fc56072d0e7c816de167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:22:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25346c9c52d249fc97302219d6bb95b6\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5af13be9051c46f0ab51425bbf4d0587 : SUCCESS in 49m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/001347950f7f4d9cb36a8cf0e499fc7a : SUCCESS in 1h 23m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8790473f601f404390519edb23c001a9 : SUCCESS in 1h 23m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2be7ab9d03724d1b84157423159049e8 : SUCCESS in 27m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f4a369541944c2eaf140895932071af : SUCCESS in 1h 01m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1c45c4a8ba8c45b5b07c66f4daf9934b : SUCCESS in 1h 09m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8274568e26234fa9b9f4cc52164665f3 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fc07e06dc184732abb3df0d292da127 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1088909962b4e829813e134280619d0 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151113f63729407cab3a67573ea63245 : SUCCESS in 6m 41s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9b22790578334c6db77ffd6c96ff3f1e : SUCCESS in 1h 12m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/97e143d7624b4ee78ec3736c18128305 : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6c31c5b607414828f955da1e240a5265daf15a78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:22:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1c463359da47cac12d707fe9b8313caabe3b6bb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:23:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21e1bd6ad6e140f394ea01658c9b6802\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70a7b4eaee1415a90083bab19c732e5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"08071f47d5ffe63368637ca76981153ac2295678","revisions":{"ac0478e374e1b96f1c29be4d2104b5225e908b89":{"kind":"REWORK","_number":1,"created":"2023-03-31 19:11:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/879177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/879177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/879177/1"}}},"commit":{"parents":[{"commit":"faf6ac9b70ce6b1e0526b4e156ed218374d14e32","subject":"CI: Fix rocky no infra wheels job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faf6ac9b70ce6b1e0526b4e156ed218374d14e32"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:16:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 19:11:00.000000000","tz":0},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nCo-Authored-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 143765fb67221cc51f1dc56a41ac2b67dddc453f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac0478e374e1b96f1c29be4d2104b5225e908b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac0478e374e1b96f1c29be4d2104b5225e908b89"}]},"branch":"refs/heads/stable/yoga"},"e67ffce05f5d80e0b646d649cc2f38fea1b60fb1":{"kind":"REWORK","_number":2,"created":"2023-04-03 11:40:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/879177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/879177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/879177/2"}}},"commit":{"parents":[{"commit":"faf6ac9b70ce6b1e0526b4e156ed218374d14e32","subject":"CI: Fix rocky no infra wheels job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faf6ac9b70ce6b1e0526b4e156ed218374d14e32"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:16:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-03 11:40:48.000000000","tz":0},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nCo-Authored-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 143765fb67221cc51f1dc56a41ac2b67dddc453f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e67ffce05f5d80e0b646d649cc2f38fea1b60fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e67ffce05f5d80e0b646d649cc2f38fea1b60fb1"}]},"branch":"refs/heads/stable/yoga"},"08071f47d5ffe63368637ca76981153ac2295678":{"kind":"REWORK","_number":3,"created":"2023-04-03 12:58:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/77/879177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/879177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/879177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/879177/3"}}},"commit":{"parents":[{"commit":"faf6ac9b70ce6b1e0526b4e156ed218374d14e32","subject":"CI: Fix rocky no infra wheels job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faf6ac9b70ce6b1e0526b4e156ed218374d14e32"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-31 11:16:50.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-03 12:58:00.000000000","tz":0},"subject":"Fix test malicious tarball fail","message":"Fix test malicious tarball fail\n\nSince I650fcbc8f773fad8116338f6fb0cf7b4f4f17b33 builds from git fails\non plugins with an exception: \u0027tarfile.ReadError: not a gzip file\u0027\nbecause the test checks only gzip compressed archives but plugins\ncreated as plain tar files. This change fixes the issue using\ntransparent compression support and also adds some debug info.\n\nCloses-Bug: #1990432\nChange-Id: If0f9b4dd058a257d0653332d1b663e150c717304\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nCo-Authored-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 143765fb67221cc51f1dc56a41ac2b67dddc453f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08071f47d5ffe63368637ca76981153ac2295678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08071f47d5ffe63368637ca76981153ac2295678"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
