)]}'
{"id":"openstack%2Fkolla~879324","triplet_id":"openstack%2Fkolla~stable%2Fxena~I14af40078508b06e46866b77ab200228ec827a0d","project":"openstack/kolla","branch":"stable/xena","topic":"bug/1970541","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-04-06 16:01:41.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-06 16:01:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14af40078508b06e46866b77ab200228ec827a0d","subject":"Add multipath to cinder-volume","status":"MERGED","created":"2023-04-03 10:51:40.000000000","updated":"2023-04-06 16:03:16.000000000","submitted":"2023-04-06 16:01:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879324-bug/1970541","meta_rev_id":"619abbebc36517afdd67087a1dfc986e5e895f06","_number":879324,"virtual_id_number":879324,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 16:01:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-04 13:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-04 13:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":2,"date":"2023-04-06 14:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":1,"date":"2023-04-06 14:33:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 10:51:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"CC"},{"updated":"2023-04-03 10:52:41.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2023-04-03 10:54:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-03 12:15:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 13:48:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-04-06 14:33:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"da898669957c56c6cc8b271a1d8aaa73414aa007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-03 10:51:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20e98ede7bd0722d2b72a5d32d886df067af5e0a","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-04-03 10:52:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f401350e2d40a8697f73c1c226bde285cc8f5337","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 10:54:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6207c206ef86b11c33d44f23dd7167d7eac29f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:15:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64ad4de617864eca8b8176f3e14481df\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/454e008e96934a64948e80bf0c96982b : SUCCESS in 30m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3648908583264040b4f59dd5ca2328c9 : SUCCESS in 27m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5182054aaef745cf877d9485f5da0386 : SUCCESS in 1h 17m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/25e5af1ff1364de7b78cb1c857628b95 : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ccd335bb30247b2af7b39f62c7f8e8b : SUCCESS in 18m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6a54d62837514b6096827ab915643b2f : SUCCESS in 16m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/87b5a780f45441e7975c20eeb69e681e : SUCCESS in 48m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76effc87f464b858e1dd027e1375adb : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e380f30a8e84eacb59a476b437940e5 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90caf3d4e1fc4288b475681f86192be8 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6e5d38b5e74db79b2d7d450602eb0e : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91fb8dcd6935406ea08180e187c1ff0a : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f25b47d37ebe47d3a696b681e9346ccb : SUCCESS in 22m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e410fd8ee4d84942ade383de5a7c3b9d : SUCCESS in 23m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c9fe77b336444a6b8b67a8432a99090 : SUCCESS in 57m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea2dc4a9296d4be6b7ca1cfd256e726f : SUCCESS in 48m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e96c7b4ba51420a875cfd2d7af0fd33 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b1aa67a17b7b459e1afed898a6832e154f9aefd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:32:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0df1b249cbf84e87ab1e660fe8a05e8d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ae70318b810428ebbe8ea1900cef159 : SUCCESS in 46m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e807be8920334bd5b86a0ffe7c4ead51 : SUCCESS in 1h 37m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb3261d014b24ce8ba16981a77b81985 : SUCCESS in 1h 30m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/573ffe7db0514a3a80e1722272f678e9 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22be38b3497f4af69418b3f3c7cf2eb7 : SUCCESS in 5m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0b3974b1097460fa881c8f71eae79cd : SUCCESS in 1h 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"593be789f50a239b9f3decf4c8361dba32f6851f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-04 13:56:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"999cbcb14beda1070b67cdb8731e57ed239f2251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 15:19:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e873e076f24e4363b74e74df762a51cb\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/589ec97affa04e6792890960b6104c2c : SUCCESS in 31m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/35aaa8a07a7d4bbc9558081fb26d72e0 : SUCCESS in 34m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3db291f0d05346c4920118ad396d31dd : SUCCESS in 1h 06m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a72ca11d02484dfd9f3f83199128374b : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/158de8a38e464ef7a85b0e1dc3b422f0 : SUCCESS in 18m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8cccf3334de944cd93bcdba9f132747c : SUCCESS in 24m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97cc397e5cb644dba75f529fb2187f0c : SUCCESS in 1h 08m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af19d91ac0b4e82a156043cd6256b97 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2300328cd04cbab2b0335245943c3c : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33f17836b7054d0d95aa8edb3cb1cce0 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2301a9258c984026b2d32f763390aa63 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d339daab83849078fbb62d0bfa4929a : SUCCESS in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/080f79a248d94a359da17a5d3aebc251 : SUCCESS in 22m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/54cce2e4a9cf4dbe83d998b3152bd8c5 : SUCCESS in 20m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a10f60fd3e74311933974a618f01ddb : SUCCESS in 59m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0d7e38c2c60437bab1190706c0d29d2 : SUCCESS in 59m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/16b824bdf42e4ccdad3f5c65dbdbedee : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":2},{"id":"930c19420a85749eebb020866442952e3e6cb622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 15:30:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da84fa36ec149b8ab84c85a5479501d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a520d33b64fb4bd5847c1cc951dfd074 : SUCCESS in 27m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d722a783c81c4bddaddd8719c40b95ae : SUCCESS in 23m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdebf6367e284cf4a14136e2c852af1e : SUCCESS in 1h 24m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f04edd3d01bb4669863bb86a8efd1d27 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/250d6ff434be4fd383cfe4a7f7809577 : SUCCESS in 3m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eea18fa6b4b84d2b832f2e02370f2d7b : SUCCESS in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9ce1a323488d81b9242ba7293a9879dc49f4337","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-06 14:33:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f39d54e15e0684fe9ff15f34d3b3d99aa0d59b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:33:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e6f7a093ee57712f574095f1f88a09422dc50cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:01:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a2a2329a3704de1973e8829fdb6f340\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1cd75390be784c998fc09ce2dc6ca30e : SUCCESS in 39m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9c3711ce24494836894b5d2c576fd1f0 : SUCCESS in 1h 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1b5c9b95154d07836d2a819818b4b7 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0de955870c4329b1fbe5195a50775b : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf5926f98774692947005d38850921e : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb87899703a449debc3d78e041884de9 : SUCCESS in 4m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f37a6686090c4ad1b203deee8690059d : SUCCESS in 18m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c3fddd0929848b9a03633912c6a3fc0 : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/55995444e81447ffa81c08b43a6d62c8 : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":2},{"id":"102a4353fce8158d3c502a3a12ffb2e6a5a8081a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:01:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"619abbebc36517afdd67087a1dfc986e5e895f06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:03:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ff47b322ed34f998d54dd9ec07ca22d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5632e39ac64a6fae51577206a7dc2e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b1c269cec7f1ba81beee33b583194c847a9662f","revisions":{"822d2e49cf1832f198f4d87785e33f9d2c1da66a":{"kind":"REWORK","_number":1,"created":"2023-04-03 10:51:40.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/24/879324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/879324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/879324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/879324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/879324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/879324/1"}}},"commit":{"parents":[{"commit":"9a86a7bf071ffa31feb3d0eaf35be31faeb20008","subject":"neutron: Add iproute-tc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a86a7bf071ffa31feb3d0eaf35be31faeb20008"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-04-27 07:54:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-03 10:50:07.000000000","tz":180},"subject":"Add multipath to cinder-volume","message":"Add multipath to cinder-volume\n\nCinder-volume container can operate on multipath devices but\nfails to do so due to absent multipath package for several images.\nThis fix explicitely adds multipath to the image.\n\nCloses-Bug: #1970541\nChange-Id: I14af40078508b06e46866b77ab200228ec827a0d\n(cherry picked from commit b9aa91381391c8aaabc7c76de39e5ee258562f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/822d2e49cf1832f198f4d87785e33f9d2c1da66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/822d2e49cf1832f198f4d87785e33f9d2c1da66a"}]},"branch":"refs/heads/stable/xena"},"7b1c269cec7f1ba81beee33b583194c847a9662f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-04 13:56:29.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/24/879324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/879324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/879324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/879324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/879324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/879324/2"}}},"commit":{"parents":[{"commit":"5da2ff0828c7376a2f16a2673cd03420da038bca","subject":"Fix test malicious tarball fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5da2ff0828c7376a2f16a2673cd03420da038bca"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-04-27 07:54:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-04 13:56:29.000000000","tz":0},"subject":"Add multipath to cinder-volume","message":"Add multipath to cinder-volume\n\nCinder-volume container can operate on multipath devices but\nfails to do so due to absent multipath package for several images.\nThis fix explicitely adds multipath to the image.\n\nCloses-Bug: #1970541\nChange-Id: I14af40078508b06e46866b77ab200228ec827a0d\n(cherry picked from commit b9aa91381391c8aaabc7c76de39e5ee258562f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b1c269cec7f1ba81beee33b583194c847a9662f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b1c269cec7f1ba81beee33b583194c847a9662f"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
