)]}'
{"id":"openstack%2Fkolla~880098","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I989cedce7a9af3882b5510d3479b29150139e861","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-04-12 14:45:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I989cedce7a9af3882b5510d3479b29150139e861","subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","status":"MERGED","created":"2023-04-12 06:22:57.000000000","updated":"2023-04-12 14:46:03.000000000","submitted":"2023-04-12 14:45:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880098","cherry_pick_of_change":878822,"cherry_pick_of_patch_set":2,"meta_rev_id":"2ad6ab60882429420b2a24b788379de57cef08d1","_number":880098,"virtual_id_number":880098,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 14:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-04-12 11:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-04-12 13:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-12 07:43:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-04-12 13:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-04-12 11:37:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 07:43:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-12 07:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 11:37:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-04-12 13:08:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c982816bfad3e19f9850e4f4d0d704ae57047f53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-04-12 06:22:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b355b147a4e82b0c0983406648ca1d9e5d1ecd13","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 07:43:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a43ae8a829a95e608c1d67dedf63dd582f184ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 07:47:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/069d12209233452697a53a0f697b46dd\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/170c082fb1324e3499b9721df3a232c2 : SUCCESS in 29m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/75e46f4e098d460aa84a0fc61098ca57 : SUCCESS in 32m 48s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f0aa8a6311f34ba0863cad3c656d5606 : SUCCESS in 31m 01s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/39eefd3ec5a7446cb344486c133ed09b : SUCCESS in 1h 04m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/63d3014db0d641bdb43dab71d1a40054 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/8d2328d67cb74f2f8a299cbeebfb2992 : SUCCESS in 1h 12m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ec8dafec2ba94b7580c75c8e53eba97a : SUCCESS in 1h 23m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/22fb506fbe204958867715c38ebc09b4 : SUCCESS in 20m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/811e48c250e0474794c1870579c462fd : SUCCESS in 22m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e0471c9b63e74b638f34968457f9993c : SUCCESS in 59m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1f443a4ee1c3448fb944bfd13b26da72 : SUCCESS in 59m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a0ed8d096046e6be221631af4ca467 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c760ab5b1eec48e9b95e7770931274a0 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e90c352fc594a688cf0ca75683fd2ab : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f834b754d904097b962e1936a520277 : SUCCESS in 7m 00s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/0d088ab84c8c494599693bb770f3520b : SUCCESS in 26m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/fa3e1085255449a1922580d5fbbaf8f5 : SUCCESS in 1h 04m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4e37de3a9d7459ba13ec3afde02069d : SUCCESS in 35m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef083554032e4e27b7f2fa7f727957de : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a820d25737b44413319e925ca3c46179f3c84329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:16:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6812a96c0f4f40c09baa62756a807c7f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cd20a6108a448b28c9867f1c90febd5 : SUCCESS in 44m 07s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/76c99ed033bd4cac8852afc82234b663 : SUCCESS in 2h 18m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a03f73642174e49bb81cf29fa197ee7 : SUCCESS in 39m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/97b118e20d0f477dbade3e63bc867c13 : SUCCESS in 2h 51m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d35e115bd624707acdf9f47a8871a79 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96ddf92c925d4946b8cf89d8125a3c91 : SUCCESS in 6m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c7772857c93c47bd90dd226d924cb78b : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64a9af2820a4edda72fb169329092c2e05881653","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-12 11:37:02.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b0129cadedeea031dd1d7770214cbad920fdb7d7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 13:08:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab7975d8be05e8b7899ebe182519ee830964dd44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 13:08:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1a609e7229aace8196eecb6fc4a6652f8b1328b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:45:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99bfd60356f41d6a3ea0834ee88a865\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b82d33ca4dc5456ba24bc05e2201476a : SUCCESS in 53m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1fb5f889e77448eea4e52467f3f7db5a : SUCCESS in 1h 19m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/164a9cb3ca414759befea41e07caff25 : SUCCESS in 1h 29m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/02d8e26972924115a99af02673f87ef8 : SUCCESS in 23m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8663fac8cf73480c83b61af4e6e72fc9 : SUCCESS in 49m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5a9f47adbdd245d8b74492ce93929529 : SUCCESS in 1h 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4183e3a381408bb7559577c5c42229 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee3b902d1cb4ac4b433fd22ee004f41 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d467b677c4f48c7b409f5c4cae2f466 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13add8e6ea3d4975b1edc799455a9a13 : SUCCESS in 5m 11s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e1244cb8fda44297b39f779566371dbd : SUCCESS in 1h 05m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e31b464dfcf4324853d9dfe57f4ba02 : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6f5f7d1aef2a84b5620a66ebf5aeb1ff8e999768","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:45:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ad6ab60882429420b2a24b788379de57cef08d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:46:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/844c27791fb644c6b93a2aec763f150b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8044564f9f29449f94ea4e6b3af459a7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb22a428acd9e3d6419b967ec92df813b6257a60","revisions":{"cb22a428acd9e3d6419b967ec92df813b6257a60":{"kind":"REWORK","_number":1,"created":"2023-04-12 06:22:57.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/98/880098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/880098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/880098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/880098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/880098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/880098/1"}}},"commit":{"parents":[{"commit":"feb5d46e63eb3cada223f21687dc94d7661e3031","subject":"Merge \"base: make curl retry on all errors\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/feb5d46e63eb3cada223f21687dc94d7661e3031"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:15:32.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-04-12 06:22:57.000000000","tz":0},"subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","message":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\n\nCloses-Bug: #2012821\nChange-Id: I989cedce7a9af3882b5510d3479b29150139e861\n(cherry picked from commit 2783fb710710103fe6aca3ff18c976590ca15e09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb22a428acd9e3d6419b967ec92df813b6257a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb22a428acd9e3d6419b967ec92df813b6257a60"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
