)]}'
{"id":"openstack%2Fkolla~880100","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I989cedce7a9af3882b5510d3479b29150139e861","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-18 11:48:37.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-18 11:48:37.000000000","reason":"Change was submitted"},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-04-18 11:48:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I989cedce7a9af3882b5510d3479b29150139e861","subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","status":"MERGED","created":"2023-04-12 06:24:59.000000000","updated":"2023-04-18 11:49:40.000000000","submitted":"2023-04-18 11:48:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880100","cherry_pick_of_change":878822,"cherry_pick_of_patch_set":2,"meta_rev_id":"7a096a97f584c7f7c86accf7e08d9d659c9c5e7e","_number":880100,"virtual_id_number":880100,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 11:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-04-13 10:46:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-13 12:56:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-04-18 10:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-04-18 10:07:02.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 07:44:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-12 08:00:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 23:12:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-13 12:56:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-18 10:07:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"ce70eb1d4c391535f11459ee20f7d00e9bebb4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-04-12 06:24:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8fe7b3141d216c602e7f149a9882f894a3d52cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 08:00:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92d22445c8f34db690a0eeb5a30b444c\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0ddd626ab364968925a15d46e6d7297 : SUCCESS in 36m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b680c68421a04fe787e6d2ed19e59c4b : SUCCESS in 35m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e7f318430e974f8bbd0a0d8ddb39e9c4 : SUCCESS in 1h 06m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0a52ef87c38f4fa9a3a9c4a7c068b932 : SUCCESS in 1h 00m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8feca452c19c45d1883013beedc348aa : SUCCESS in 1h 05m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5a208107f49e4d2aade0c91202173631 : SUCCESS in 24m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac74544d85904c8396e8e5b2ce96f72c : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a54713febea4e4599fa8e03589c77e1 : SUCCESS in 51m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cdc1499ca8448089fa94bc37adc888 : FAILURE in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97349923fab7476eb5d9168c2d44f771 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b54b19d1014454ebd3d56f41bc9ef43 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf7b0c94a5664bfcbba751fd394bf4e3 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68b95b680b149ebaf925fdd71cfc9a3 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbe2935dab2d414fa0cc8d15a28a0625 : SUCCESS in 20m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebf9f7ddd5bf4f39a2cb8122d858fa5b : SUCCESS in 26m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1df0aeb88fa44762a0903186e5965127 : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f31bbe6f8bd4d25881c072fc1c707d7 : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d221e2f3cd554fafab970a71afde9bf3 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f8af38ef74b12d3619f2726d453a82e414b44fa3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 09:28:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75dc4ab68b8343739a6766d6db71764b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c450e29e7dad469f9a673bad678d4e45 : SUCCESS in 2h 26m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50dbf29c227b455b94320b35c43cd91d : SUCCESS in 43m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/042aad09ad0e4a438808a512cdff85b3 : SUCCESS in 3h 01m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/315e6335616c4f6e87312d26461d1be8 : SUCCESS in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf5ca9a673402b3c2d91a3ab12b9f38b87b1f791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 21:49:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"498e0c77582bbf53bca25ccba9fac1ed0863aba3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 21:51:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"42e3af27e7668ac77f14762be7d186995a536f14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:57:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b27af788a24abd8da66c6ba567d2c8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d103567ce4924a0a88eb3ab5888f5d79 : SUCCESS in 24m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6016524750d14d9fbe7b547cbca907dc : SUCCESS in 15m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3037e54f095482e80b4ace4a81619bc : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b50fc2a89056435caa99787a4ca472b1 : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"712757abb028a003b7e461d61d34de8e7594e572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 23:09:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94e894d8285443fab2a0b0b10395568b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7f59b9a08ee64238b8b0bcf3a4fc18d6 : SUCCESS in 40m 59s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a25d614947be41a29b35278f86424654 : SUCCESS in 40m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9412b740c6a246afbd626908078286c1 : SUCCESS in 1h 07m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6852d1910a3f4542a5c45fb7d09e7e95 : SUCCESS in 1h 01m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b0a434304620482983e293430ceb7600 : SUCCESS in 1h 17m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/65063e5722504d318935697affcf12ce : SUCCESS in 20m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fbf7da707ca0425daea6b36bf1f69172 : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa79dccc085e4e0b91c2ac136d2d1a31 : SUCCESS in 57m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a24a49efe2cb407dba441a2f966946f0 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e68319bb4a1944df8e669e07fbacb0df : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/197205b157c245cd95294664c53be897 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32c3aa23d15044a794f97886ea315f91 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85af6f1527704e0f99317dfbba2474d4 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acb95cfa225446489cba39eaf33897b1 : SUCCESS in 23m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebd90d0abaa8422a8aaa1ea196f5e678 : SUCCESS in 21m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9daf6de9c2947d1b5ee099442dbb248 : SUCCESS in 1h 00m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59be9ef5ac2f4b45a8291579d17c104f : SUCCESS in 50m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dcad68157a6f4a93a8c68fedfb2bdbf2 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":2},{"id":"82fd982ed020bd2c95dd119d2e29bde9cefb2f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 10:46:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"31d892024c769e0324f682a7cf3ed10745b788fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 12:07:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bafd8760e0448e0bafe0cbe4f9b2627\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f541da28ec07437aa770ba5f3c24ad58 : SUCCESS in 44m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9005d7df6ce483db91d729de8e0dfaa : SUCCESS in 37m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/09005b1d1ba24d74911082b8416bdea8 : SUCCESS in 1h 18m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f38ccb5579547789dfc45073a78f9b3 : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4806fe526ff56c827e8a3b7d2c6a430b875e94f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 12:13:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a58f9fa7f50b426592f67293c49fb081\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d2f42deb0f544d1ca0d1cc253042eae5 : SUCCESS in 30m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f24eb2dfcb404806acf51c26340c3dfe : SUCCESS in 27m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/62f9fc70e77e4c6caebe9b7fdeaec871 : SUCCESS in 1h 05m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a49e1ca2b2e4032b67391192ed09875 : SUCCESS in 1h 05m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/36cbed7971d845a29513bbf709b74f8d : SUCCESS in 1h 16m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9b52ead0fd21437193eae7c5ab2f9dd2 : SUCCESS in 19m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5f11feec63be4011906ac2605de7833f : SUCCESS in 20m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/018666ac1ab04e7798a84c3e303216de : SUCCESS in 55m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f06b7482fd643deb438b76ce0a4fa5f : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c482d70ffc5142fca127405b3cdbfa1e : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2a9af218e6d458da9d245d61ecbc6f4 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5173c7564d1f454fa42d7f7826606ed6 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f66fef20659f4533a675e4513236c2b4 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72bd7370c99b4657851b26ee4192b79c : SUCCESS in 19m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5e9b320b1e443508681c71b96155acf : SUCCESS in 21m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23069435b6c6418a9463487f5b663988 : SUCCESS in 51m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6dd519aaf5a40afad1b074ac4ec72a2 : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c12263c80a8549cf8472479f084ad419 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":3},{"id":"bdc537ea8e3bda90c8167c73f0e9531158189177","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-13 12:56:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a78efa5e1aff4e6ac6fe61b097d2864c97d16da5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-18 10:07:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e470f69041fcf5becee1d895375004b9fe391309","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 10:07:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7af18d3e3b9e026bd549a4d5511a35cb669c88b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 11:48:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2adeba5137d411f8407d5f084f12a83\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/03af73ee814d45a28094d5d26ce0cc7a : SUCCESS in 31m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2cbde735d00a487b814bf943ceb58808 : SUCCESS in 1h 09m 37s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/448763ddf0b4480380527b99bd041f6e : SUCCESS in 1h 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b80b53c2f94d16b28e7b8b92d1e04e : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7280c07d195c46bd8eba1ed5471b2fc5 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53d82c38378843628196cbd21bb2817e : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f80a7c1a4d4e639b4be5a61c147de9 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a57092bab0442f58b0844d6fb74a5ef : SUCCESS in 20m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76a53fb33a0c4b5b819b64f00bf73cbf : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/44ad3532fd7043e4aaaca1d2c7c62e17 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":3},{"id":"c632ebb4f06d957fb571cfa5d4c4ccec3591d20f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 11:48:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7a096a97f584c7f7c86accf7e08d9d659c9c5e7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 11:49:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0445113ce9e44965a1e98fe3a81aaa72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43dfe279f97f4a65ac47a89e423a6f74 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b0574dcec2059c333622ce1e23e914c1c5c1f3e","revisions":{"5d69d958a76159d8c91a45ed0f0580c4171d22ae":{"kind":"REWORK","_number":1,"created":"2023-04-12 06:24:59.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/00/880100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/880100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/880100/1"}}},"commit":{"parents":[{"commit":"e9f0a9aaea1616c62a4014957e06c3b61fd9c50c","subject":"Merge \"CI: Retry docker push on publish jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9f0a9aaea1616c62a4014957e06c3b61fd9c50c"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:15:32.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-04-12 06:24:59.000000000","tz":0},"subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","message":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\n\nCloses-Bug: #2012821\nChange-Id: I989cedce7a9af3882b5510d3479b29150139e861\n(cherry picked from commit 2783fb710710103fe6aca3ff18c976590ca15e09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d69d958a76159d8c91a45ed0f0580c4171d22ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d69d958a76159d8c91a45ed0f0580c4171d22ae"}]},"branch":"refs/heads/stable/wallaby"},"8b01bce205bb0942e31b83cbb57c7ac36e23085f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-12 21:49:59.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/00/880100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/880100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/880100/2"}}},"commit":{"parents":[{"commit":"9b350df4b3188183914f96b44b633fab3ad9afa3","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b350df4b3188183914f96b44b633fab3ad9afa3"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:15:32.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-12 21:49:59.000000000","tz":0},"subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","message":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\n\nCloses-Bug: #2012821\nChange-Id: I989cedce7a9af3882b5510d3479b29150139e861\n(cherry picked from commit 2783fb710710103fe6aca3ff18c976590ca15e09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b01bce205bb0942e31b83cbb57c7ac36e23085f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b01bce205bb0942e31b83cbb57c7ac36e23085f"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"2b0574dcec2059c333622ce1e23e914c1c5c1f3e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-13 10:46:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/00/880100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/880100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/880100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/880100/3"}}},"commit":{"parents":[{"commit":"779aa83f4448df2ad4db7309cca496301b777ff8","subject":"Fix test malicious tarball fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/779aa83f4448df2ad4db7309cca496301b777ff8"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-03-29 09:15:32.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-13 10:46:56.000000000","tz":0},"subject":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package","message":"nova-libvirt: Fix for missing libvirt-daemon-driver-nodedev package\n\nCloses-Bug: #2012821\nChange-Id: I989cedce7a9af3882b5510d3479b29150139e861\n(cherry picked from commit 2783fb710710103fe6aca3ff18c976590ca15e09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b0574dcec2059c333622ce1e23e914c1c5c1f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b0574dcec2059c333622ce1e23e914c1c5c1f3e"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
