)]}'
{"id":"openstack%2Fkolla~880154","triplet_id":"openstack%2Fkolla~master~I9a8703861e174d8933e0a2cce63b076145eaa732","project":"openstack/kolla","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-18 10:06:24.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-09-20 13:14:09.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9a8703861e174d8933e0a2cce63b076145eaa732","subject":"WIP: we want to track github releases easier","status":"NEW","created":"2023-04-12 13:51:54.000000000","updated":"2024-12-18 21:25:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"137129d652933eb7de4b50dade139577d4297ac3","_number":880154,"virtual_id_number":880154,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-18 15:46:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"reviewer_updates":[{"updated":"2023-04-12 15:02:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-12 15:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 10:06:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-20 13:14:09.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"2c82416ca5ac6f3bc933a4abe42ae72be39b1c93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-12 13:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c0ae1ea692f50910fdee8aae827b04c4a813852","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:02:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0a549a7d9841f8a87ec87d2be169f6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/52fa0e24483c4314bfb12b2f99347526 : SUCCESS in 19m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/40ffc082aa324c2faf0b51e31fb4e526 : SUCCESS in 16m 43s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee64ea3dbd59499c9506c0f18e8140a0 : SUCCESS in 1h 08m 39s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/274ff84d5e2f45f4822cb0e8c877808b : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cfeaaf3eb053429cbe7dd9be0c0fe7ac : SUCCESS in 4m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f0eece5597ab4d2db96967d43ca03d31 : SUCCESS in 16m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c663c122128451cb0f653597117ac72 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e94eff1e60c91a4dd3ac45990e8b022fdf9331ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:29:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8073367d56184f8386dac2df6bd9ed38\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5efcf6cafe5b4618b3f6757f38e3633d : SUCCESS in 16m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/da915549b6c2427c8c4d7455fee415be : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/75ae5f84fa1647029f22dcf99cb1ffd6 : SUCCESS in 18m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c73a02ee27d74eba9255a4345a8835ac : SUCCESS in 1h 07m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6f1a13d36dd4851a116090beeb608ff : SUCCESS in 1h 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930a977bcb7447b6b4b9f37b08819a63 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e75d097ec5a46269ca7062a86a33f7c : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8200dfb5ff994a019b89528b0e6da448 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06c314f203ef4682a0bc6d0007a40971 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c3a2a645264246b2074f6c4da515f8 : SUCCESS in 4m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/74ef0102fe484ecebfeb1c143c578ede : SUCCESS in 19m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d14760e2837a46888b2e8a884ea14218 : SUCCESS in 1h 14m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0cbd412d11944b7693d0dcd3b12c73b4 : SUCCESS in 1h 31m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d7c589e61d7448da844a7fab0e2af4d6 : SUCCESS in 24m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/828b84c326a746bc8cef085ffacf4b53 : SUCCESS in 56m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cacb06fc6f664cfe9d5c31a1fe4c3421 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ed4cd1ef04968c51955d7c5ea13ed7fec0eb0ca2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-05-02 12:42:44.000000000","message":"Topic github removed","accounts_in_message":[],"_revision_number":1},{"id":"4555501a51cedfd770f689a697eb49df1e7504e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-18 13:28:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_24072\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"_revision_number":2},{"id":"de09093dad318f06a7070c0000be8feceb262002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 15:46:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9382646d0894b6584617213beb41f51\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/49a481f8ff25428589c6bb210921ab63 : SUCCESS in 31m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fb605a043af64b3f826babb54be172fd : SUCCESS in 1h 04m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a57f969bed674420a732ba5091d20b27 : SUCCESS in 21m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/385267a4e6fb443785a87ab9fb3b2605 : SUCCESS in 39m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bb2b7a6d3f6c46d68b3a754ea1b703e0 : SUCCESS in 1h 11m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/001e3ed5f2c048d0a1cefdb778d3754e : SUCCESS in 1h 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe4a2ab6337483087b8dd142c771aff : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e5658759c844e3da5fc38f6dedaad6a : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b897cec129b44f8994465e782379260c : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/175b062b75ec4897afdb10170f2f0f7f : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696580044e9c4c8ba91787c38b352f88 : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36f0fc17966e4d03affd2cdd408e4e12 : SUCCESS in 31m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cc9bad2bd70e448b931fc12a37bd3661 : SUCCESS in 39m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5571ff7004ea48d5a22ad3872b290751 : SUCCESS in 59m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8ade9c01d43d439b95f5d0ba55a58471 : SUCCESS in 1h 34m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/97b8d519c03a4a41bf45ac6798c49739 : SUCCESS in 24m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9df76b1645624da8944356db0c836dba : SUCCESS in 19m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/851794758c1649a2b03cdf197f70c635 : SUCCESS in 1h 12m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/56fb4bedaf8e44ffbd1aaaf02255ff86 : SUCCESS in 1h 56m 25s","accounts_in_message":[],"_revision_number":2},{"id":"137129d652933eb7de4b50dade139577d4297ac3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 21:25:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96f3532b2e2b45219a5f7a54133bd3ac\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e6b663be61d84199af2e01e6f602eb12 : FAILURE in 2h 38m 19s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2c865e32d82340f19080937e30cf571f : FAILURE in 2h 36m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/623949c141de4ee6a3d1933a955d43e6 : SUCCESS in 2h 38m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be456960b19e4ae7b35a827d6654f1ef : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7da0327b7958483282f13a3c1a79a327 : SUCCESS in 5m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4e27a3510b114fcf9eaf621eff498f53 : FAILURE in 2h 44m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a5ce229781ee40e98f034d5342e5aff4 : TIMED_OUT in 3h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6438da7e70cf30ceb8a54b1eca47b7cfac3046c2","revisions":{"c4558bc4a305caf21d1c3d892296769e890ea2a2":{"kind":"REWORK","_number":1,"created":"2023-04-12 13:51:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/880154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/880154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/880154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/880154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/880154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/880154/1"}}},"commit":{"parents":[{"commit":"be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f","subject":"Merge \"neutron-mlnx-agent: pin pyzmq at 21.*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be0e89d600ddc6b20bb96cbd5f4a2ec899b7051f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-12 13:50:13.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-12 13:51:45.000000000","tz":120},"subject":"WIP: we want to track github releases easier","message":"WIP: we want to track github releases easier\n\ntools/check-github.py is start of something. All it does for now is\ngoing through Github API and check for latest releases of components we\nuse.\n\nRate limited.\n\nChange-Id: I9a8703861e174d8933e0a2cce63b076145eaa732\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4558bc4a305caf21d1c3d892296769e890ea2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4558bc4a305caf21d1c3d892296769e890ea2a2"}]},"branch":"refs/heads/master"},"6438da7e70cf30ceb8a54b1eca47b7cfac3046c2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-18 13:28:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/880154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/880154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/880154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/880154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/880154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/880154/2"}}},"commit":{"parents":[{"commit":"b1be32617803e2a6c24f449a37622717e2a6d102","subject":"Merge \"magnum: Add helm binary for magnum-cluster-api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1be32617803e2a6c24f449a37622717e2a6d102"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-04-12 13:50:13.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2024-12-18 13:28:27.000000000","tz":0},"subject":"WIP: we want to track github releases easier","message":"WIP: we want to track github releases easier\n\ntools/check-github.py is start of something. All it does for now is\ngoing through Github API and check for latest releases of components we\nuse.\n\nRate limited.\n\nChange-Id: I9a8703861e174d8933e0a2cce63b076145eaa732\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6438da7e70cf30ceb8a54b1eca47b7cfac3046c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6438da7e70cf30ceb8a54b1eca47b7cfac3046c2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
